)]}'
{"id":"x%2Fpackstack~931501","triplet_id":"x%2Fpackstack~master~Ibb1a073ba599d8d052fef1e4bc77cf99ac66124c","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-10 18:51:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibb1a073ba599d8d052fef1e4bc77cf99ac66124c","subject":"Select correct memcache server value according to IP version","status":"MERGED","created":"2024-10-04 15:47:09.000000000","updated":"2024-12-10 18:51:18.000000000","submitted":"2024-12-10 18:51:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931501","meta_rev_id":"39d933644d64881067bf124553f8f74ed863dd74","_number":931501,"virtual_id_number":931501,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 18:51:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2024-11-18 06:56:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-10 17:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-10 17:29:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-04 16:58:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-07 07:13:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-12-10 17:29:29.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"859af2e6d6450a1384134fb9b770131c4ad9c36d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-04 15:47:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c7fdd4a16231a9ac01ee1eedde6225f781fe8c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 16:58:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74e085758150478994911574af5f4695\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57637c4b4a304b1cb63d0bfb70049c5a : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77251d61fd1448f2844f13f5c8e4c81f : SUCCESS in 2m 41s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fe06b5d1e3434f7db56bc7f5d12f7a54 : FAILURE in 23m 06s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4cb212c48dac403e816338a0d6f4a712 : FAILURE in 11m 21s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5bb0a7d4eec242d5bbf02bec63e7e38a : SUCCESS in 1h 04m 29s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2f2b4ad10fff42629e432daa73b77996 : FAILURE in 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"30133817ebe9422b794be41e9f709062ef9a46fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-04 17:21:48.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4f49c43ff81241b29fff179ed779d02c121b208b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 18:21:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80357f219e6348928e889a5bec1107e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8023dbd80dd4672ac5880b1dee6a43d : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd58de1b1c3245d892f1c1ba4d4e3120 : SUCCESS in 2m 46s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e42f3d1fd70549b18910196446a11507 : SUCCESS in 53m 07s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/85d4766da770494bbc18b546ae14f632 : FAILURE in 21m 15s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9b528e4e2363444ba4ccedb7c0335d22 : SUCCESS in 33m 13s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1c456daac5e64498ab6fd891c0ba56ac : FAILURE in 24m 20s","accounts_in_message":[],"_revision_number":2},{"id":"dd58a24769b9553ca0f9b6b651da8749bd49cda1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-05 07:28:04.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3fa85a6728c964bc59e0d8fc10d3bc38297ced75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 08:32:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c26a3fc9d48e47b59cfc3db8490ca2b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d879079abdc6440cab1adf9ea13be5fa : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a2d927a65e94140aeae2eed87346c88 : SUCCESS in 3m 11s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/117316cd489f4ed9a3cd20a8b17bf4dc : SUCCESS in 30m 30s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/73ce4aa113e74104b2cd252907636f14 : SUCCESS in 29m 21s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3a38402852e94538ad9cf626970fdd80 : SUCCESS in 58m 43s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/234441214b12455e81a6657874993f00 : FAILURE in 21m 50s","accounts_in_message":[],"_revision_number":3},{"id":"8fd819b3b9ad193e2e217622d8249f103a9a90a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-05 08:54:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1c6f09bb54f18cd2353beb69a67be4f2ea20f5da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 09:57:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3741aff187d43c1a05dd629502065d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1fdf4df15ff451db003eaab86acd965 : SUCCESS in 2m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15928c2e7c3442468fceb5e718027131 : SUCCESS in 2m 58s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e96d760158fe49f69aed5a1a59d8a57e : SUCCESS in 28m 23s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b1c9c51efe8e49fdb03821d3cf736143 : SUCCESS in 52m 46s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d22d4a9796e04172ba44b8166725efb1 : SUCCESS in 32m 35s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e1d933f43c8c4f30befe4092de446828 : SUCCESS in 56m 49s","accounts_in_message":[],"_revision_number":4},{"id":"4f4785c04bdad1b4206a059d255023fa5cb53fc4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-07 07:13:32.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"93eccd5f5e63627b1555e2aea3a39d2c847dd2ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 06:56:37.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"44676893d0a661cd6a8895cd8a3b35502e0400ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 08:06:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b44c62857e0468c92e7010920a8ef82\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e218bf7e2ff475fb4ed0e381a4b5f91 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4daf45668a834e1fb5cdf3f159ef6a25 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2963159761124c8bb6b01d273e454788 : SUCCESS in 3m 07s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b2a9392a92c74f0886e39dd6c5f7ed34 : SUCCESS in 54m 29s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bdbadb64fe644687933bc17bf0e82bf2 : SUCCESS in 58m 21s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/01f3d481ce38475f88d18b4ff70a2b0f : SUCCESS in 31m 25s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/aceacb66727f4282bb4aee302b6703a1 : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":5},{"id":"59187d2ba79e81e1dc7824cb0d3e728e756b3200","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-10 17:29:29.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f38226ba06b4a25415f9e0af26d45fd42d633ef7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 17:30:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c9041cccbbf7c5a857855997b20fe1667a72ec36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 18:51:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77d40c55cd9448bfad54750cb5a14d0c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c7a69f9b074bb4b9bbefa998aa3bd4 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abec329b8fe74b71b261d1c9db62a4ef : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea321b155b0545618d3288aa76592fb6 : SUCCESS in 3m 38s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a886d11f4e2441e19129c3a908e12b23 : SUCCESS in 42m 10s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b6d424151a2b40d2b322eb3b36d24f1a : SUCCESS in 41m 08s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/35ab186eaa1b46a99d853069a5e9c07e : SUCCESS in 46m 06s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/87d050d48adc4156acd4ffd3f8644022 : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":5},{"id":"39d933644d64881067bf124553f8f74ed863dd74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 18:51:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e82988b461a24c6d2436a27d86336db8901b1ebe","revisions":{"7aff2013c3573c6964ed25ccbf006f5addd43fa9":{"kind":"REWORK","_number":1,"created":"2024-10-04 15:47:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/931501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/931501/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/931501/1"}}},"commit":{"parents":[{"commit":"5419da93b85b48671c6e36426deb1d63186a0521","subject":"Merge \"Bump cirros version to 0.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5419da93b85b48671c6e36426deb1d63186a0521"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-04 15:46:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-04 15:46:37.000000000","tz":540},"subject":"Select correct memcache server value according to IP version","message":"Select correct memcache server value according to IP version\n\nChange-Id: Ibb1a073ba599d8d052fef1e4bc77cf99ac66124c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7aff2013c3573c6964ed25ccbf006f5addd43fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7aff2013c3573c6964ed25ccbf006f5addd43fa9"}]},"branch":"refs/heads/master"},"fa95f270b892dc009025cc9ed6f9d62e322e1943":{"kind":"REWORK","_number":2,"created":"2024-10-04 17:21:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/931501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/931501/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/931501/2"}}},"commit":{"parents":[{"commit":"5419da93b85b48671c6e36426deb1d63186a0521","subject":"Merge \"Bump cirros version to 0.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5419da93b85b48671c6e36426deb1d63186a0521"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-04 15:46:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-04 17:21:48.000000000","tz":0},"subject":"Select correct memcache server value according to IP version","message":"Select correct memcache server value according to IP version\n\nChange-Id: Ibb1a073ba599d8d052fef1e4bc77cf99ac66124c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fa95f270b892dc009025cc9ed6f9d62e322e1943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fa95f270b892dc009025cc9ed6f9d62e322e1943"}]},"branch":"refs/heads/master"},"3eb64e043ed5b857818ec29b26d555bf9b284a70":{"kind":"REWORK","_number":3,"created":"2024-10-05 07:28:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/931501/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/931501/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/931501/3"}}},"commit":{"parents":[{"commit":"5419da93b85b48671c6e36426deb1d63186a0521","subject":"Merge \"Bump cirros version to 0.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5419da93b85b48671c6e36426deb1d63186a0521"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-04 15:46:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-05 07:28:04.000000000","tz":0},"subject":"Select correct memcache server value according to IP version","message":"Select correct memcache server value according to IP version\n\nChange-Id: Ibb1a073ba599d8d052fef1e4bc77cf99ac66124c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3eb64e043ed5b857818ec29b26d555bf9b284a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3eb64e043ed5b857818ec29b26d555bf9b284a70"}]},"branch":"refs/heads/master"},"a804d567362fa67ff127e5d2172292b8ab5b2954":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-05 08:54:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/931501/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/931501/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/931501/4"}}},"commit":{"parents":[{"commit":"2dfe0d24631ff10812fc70bcb35f726f19c2409d","subject":"Merge \"Consistently configure authtoken options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2dfe0d24631ff10812fc70bcb35f726f19c2409d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-04 15:46:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-05 08:54:27.000000000","tz":0},"subject":"Select correct memcache server value according to IP version","message":"Select correct memcache server value according to IP version\n\nChange-Id: Ibb1a073ba599d8d052fef1e4bc77cf99ac66124c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a804d567362fa67ff127e5d2172292b8ab5b2954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a804d567362fa67ff127e5d2172292b8ab5b2954"}]},"branch":"refs/heads/master","description":"Rebase"},"e82988b461a24c6d2436a27d86336db8901b1ebe":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-11-18 06:56:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/931501/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/931501/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/931501/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/931501/5"}}},"commit":{"parents":[{"commit":"120c142c1dea4cca18fe7518a0c12f653067ae93","subject":"Merge \"Create volume types by built-in logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/120c142c1dea4cca18fe7518a0c12f653067ae93"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-04 15:46:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:48:15.000000000","tz":540},"subject":"Select correct memcache server value according to IP version","message":"Select correct memcache server value according to IP version\n\nChange-Id: Ibb1a073ba599d8d052fef1e4bc77cf99ac66124c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e82988b461a24c6d2436a27d86336db8901b1ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e82988b461a24c6d2436a27d86336db8901b1ebe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
