)]}'
{"id":"x%2Fpackstack~931629","triplet_id":"x%2Fpackstack~stable%2F2024.2~I0cb4d6d0f5ae8d77c4e551cd2ac1154bc9a62278","project":"x/packstack","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-07 15:56:03.000000000","reason":"Change was submitted"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-10-07 15:56:03.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-10-07 14:21:24.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-10-07 15:04:52.000000000","reason":"\u003cGERRIT_ACCOUNT_31068\u003e replied on the change","reason_account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},"34750":{"account":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"last_update":"2024-10-07 15:56:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0cb4d6d0f5ae8d77c4e551cd2ac1154bc9a62278","subject":"Switch to stable/2024.2 branch","status":"MERGED","created":"2024-10-07 09:50:08.000000000","updated":"2024-10-07 15:56:03.000000000","submitted":"2024-10-07 15:56:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931629","meta_rev_id":"da38e5b2fec703603452c71f9d06359e93545f7e","_number":931629,"virtual_id_number":931629,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-07 15:56:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-07 14:21:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-10-07 14:21:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-07 15:04:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-07 09:51:22.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"state":"REVIEWER"},{"updated":"2024-10-07 09:51:22.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2024-10-07 09:51:22.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-10-07 09:51:22.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-07 10:16:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44e191f780fad641ac1fc33f28fd101d98f3b053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-07 09:50:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9b26e4f9eb1f839476b86f01b15b48bdae67b74","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-07 10:00:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67ac0221d81512d1ed5c93aaa5e287850c837ed4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 10:16:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6325c0fdb974a41b8e8b7f5b832ccfe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f031c269caf1491c9a076b585351d78f : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84e6f5696d75491cb163b60baa47bbe3 : SUCCESS in 3m 50s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/41d21bdd5f844207865bb47cb893f629 : FAILURE in 10m 02s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ebb23eb12caa41af816587d4a55a859f : FAILURE in 8m 19s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/fe09aff2289a47de9cf95841cf032025 : FAILURE in 19m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3aca0d39962245f1908243ae9f68b3a3 : FAILURE in 21m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d2f07d4e8b7cc584911a7edf18f261ba0d527384","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-07 12:14:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ca4e5db2ec6941b60ccaee9ccbb6192a96156a6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 12:51:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b041c9ae5c034578a7e65e3e36edcc5e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d08a3b5a736b4457b698e2dbb9e4ec1b : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0caba90c4c484c27bb1805e9bc074ab5 : SUCCESS in 2m 59s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b1a32d59cae641368ae55e404564fdc8 : FAILURE in 25m 35s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/97c09e30e1a649d690b9892c5e8eaf02 : FAILURE in 24m 55s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/983f9e88674a4f81893eeb23e17a5e75 : FAILURE in 23m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d31cdd2197414e6a8b685aa76ef7a63e : FAILURE in 24m 38s","accounts_in_message":[],"_revision_number":2},{"id":"cc20f0e3a730dca24f0c871d23a4023b10610091","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-07 13:19:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7b2daf5ae825548e657addb1b94e2d1c7a4689d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 14:07:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37c0b3c4cd624f65b4acf33738ba208c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3194a9326f4042b8c24792cc9dbc3a : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c797a23dab043bcad7971aa143dcc84 : SUCCESS in 3m 44s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/461dbfb5497741cb99c656b2e3c9fe9b : SUCCESS in 43m 23s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9313fd3559ae47dfbaa0de1da77c026a : SUCCESS in 35m 57s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/acda98df44384484b9f4e3b576e46146 : SUCCESS in 47m 21s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e7a4ed7f37b24842b8a27a7e8547336e : SUCCESS in 40m 52s","accounts_in_message":[],"_revision_number":3},{"id":"1d0332efa13f6fcf73010febd9c4a7975c45c30d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-07 14:21:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bd4c96e84f61dc19b3344961a133e3bd624bc8b2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-07 14:21:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"49796fb46db326dbbd1bc569a96bb48bbfc22197","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-07 14:21:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c83dc7e186db7704ee9c79a5161c0606089e529a","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-07 15:04:52.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3a3fb19ba056010a597be6b8c032b5b7820cd1de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 15:05:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ffdebdb71d10d6c6d1dfbc18c2897b5e5c6230c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 15:56:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7fb6a0863e947529e26ed8dfb06eae3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16c7879f4f9e43348fa29b53c1314612 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae6caa8fff404851851455f26714c3fa : SUCCESS in 3m 51s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/afc17b1c88574c67b9cd1542dc54fca7 : SUCCESS in 34m 03s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/584bc2e253d2440d842383c7760fd2c2 : SUCCESS in 31m 17s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/98450fc4dfc14d198b7b10aeaca1c09f : SUCCESS in 36m 55s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d964532e8f904af99e21359e6b0d3483 : SUCCESS in 49m 03s","accounts_in_message":[],"_revision_number":3},{"id":"da38e5b2fec703603452c71f9d06359e93545f7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 15:56:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2c587c1bc81349e15c1aed37aa3c48f21ea22008","revisions":{"945180b48df3848d556038f944f4597c7ce6d106":{"kind":"REWORK","_number":1,"created":"2024-10-07 09:50:08.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/29/931629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/29/931629/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/29/931629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/29/931629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/29/931629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/29/931629/1"}}},"commit":{"parents":[{"commit":"2dfe0d24631ff10812fc70bcb35f726f19c2409d","subject":"Merge \"Consistently configure authtoken options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2dfe0d24631ff10812fc70bcb35f726f19c2409d"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-10-07 09:42:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-10-07 09:42:31.000000000","tz":120},"subject":"Switch to stable/2024.2 branch","message":"Switch to stable/2024.2 branch\n\nChange-Id: I0cb4d6d0f5ae8d77c4e551cd2ac1154bc9a62278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/945180b48df3848d556038f944f4597c7ce6d106"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/945180b48df3848d556038f944f4597c7ce6d106"}]},"branch":"refs/heads/stable/2024.2"},"fb1f620eda37a36afe2f68e3af814fd13e3b102b":{"kind":"REWORK","_number":2,"created":"2024-10-07 12:14:05.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/29/931629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/29/931629/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/29/931629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/29/931629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/29/931629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/29/931629/2"}}},"commit":{"parents":[{"commit":"2dfe0d24631ff10812fc70bcb35f726f19c2409d","subject":"Merge \"Consistently configure authtoken options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2dfe0d24631ff10812fc70bcb35f726f19c2409d"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-10-07 09:42:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-10-07 12:14:01.000000000","tz":120},"subject":"Switch to stable/2024.2 branch","message":"Switch to stable/2024.2 branch\n\nChange-Id: I0cb4d6d0f5ae8d77c4e551cd2ac1154bc9a62278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fb1f620eda37a36afe2f68e3af814fd13e3b102b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fb1f620eda37a36afe2f68e3af814fd13e3b102b"}]},"branch":"refs/heads/stable/2024.2"},"2c587c1bc81349e15c1aed37aa3c48f21ea22008":{"kind":"REWORK","_number":3,"created":"2024-10-07 13:19:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/931629/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/29/931629/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/29/931629/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/29/931629/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/29/931629/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/29/931629/3"}}},"commit":{"parents":[{"commit":"2dfe0d24631ff10812fc70bcb35f726f19c2409d","subject":"Merge \"Consistently configure authtoken options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2dfe0d24631ff10812fc70bcb35f726f19c2409d"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-10-07 09:42:31.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 13:19:07.000000000","tz":540},"subject":"Switch to stable/2024.2 branch","message":"Switch to stable/2024.2 branch\n\nChange-Id: I0cb4d6d0f5ae8d77c4e551cd2ac1154bc9a62278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2c587c1bc81349e15c1aed37aa3c48f21ea22008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2c587c1bc81349e15c1aed37aa3c48f21ea22008"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
