)]}'
{"id":"x%2Fpackstack~931918","triplet_id":"x%2Fpackstack~master~I028178976b96199cd773022f5b8fec70cf223bde","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-09 18:20:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I028178976b96199cd773022f5b8fec70cf223bde","subject":"Adopt to latest Puppet OpenStack modules","status":"MERGED","created":"2024-10-09 13:30:26.000000000","updated":"2024-10-09 18:20:06.000000000","submitted":"2024-10-09 18:20:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931918","meta_rev_id":"92cdf563e1e46958f94ac17cb91fa06f6e2bda36","_number":931918,"virtual_id_number":931918,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-09 18:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":2,"date":"2024-10-09 14:52:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":1,"date":"2024-10-09 15:04:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 14:43:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-09 14:52:13.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"713761132ec664cc34d84a523491638c2999c553","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 13:30:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5527809518545528e482aacb56a036efa89d3e2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 13:32:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"52b3a051abe74f6a0a79551237ae111965586669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 14:43:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b05c15d2616c4f5493279ca515490592\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed36ff7027e14825bfe8ef8718e18555 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d72b78ff9fa948fcaf47d3404e56ab65 : SUCCESS in 3m 18s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a54eb9459e494814a6443013215858d9 : SUCCESS in 54m 19s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7ee1d34f24204fbab4ad5f8520573679 : SUCCESS in 51m 38s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8378949715784241b96ce23ac2028e19 : SUCCESS in 1h 05m 26s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/530403ae5b7248b6b382955619ec155b : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":2},{"id":"5f5e2382deeb4b164c9ff61db87ec7a3a6e46987","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-09 14:52:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a7a0dc4dde2ef246aef4f0688a9962692abe5b69","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-09 15:04:19.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fcc392c7e3ccb075b4a5639eaaa149be5b3b5877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 15:04:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"92ac90eb1a23c8a7b285aa2fffc78ab68abbb8ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 16:08:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f326c64978d4368a2c761c898a7dde2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa409a666254c05975fd98a54e614ae : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e107716a91541a8bb4545c17830e26c : SUCCESS in 2m 49s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/955e811fccac49659847a0e6672bb836 : SUCCESS in 58m 30s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b889aef565674d21954e8b902d33a1a5 : FAILURE in 16m 05s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1526c8c8e7ce4b4bbb91ef237441fc88 : SUCCESS in 56m 41s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/873714cb39c849cfb2f59e0b5aac8362 : FAILURE in 30m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f5800a267f8bc1ace038f4766cf9b8db3d8fca9c","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-09 16:13:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b88b0d2bcdce46a694f22a1513be5ac37515152","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-09 16:14:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"98d6dd057980a883fb78dc1d3a2392e722be4119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 17:12:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e73de6f6770f4e89bcf5e33b36b9b589\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b982fef5b894e7196065d44fcb1cff3 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e06d355e7c84a62910a2fe6a928548a : SUCCESS in 2m 48s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7c0b3c36060944a789276d9785ea9307 : SUCCESS in 53m 10s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7f99aae7bcc64168be2237caddda609a : SUCCESS in 53m 21s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0572c833c94b42ea94d7cf32cbcb115e : SUCCESS in 53m 12s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c490ac0e60f048ebaa6c82718c1e9664 : SUCCESS in 46m 47s","accounts_in_message":[],"_revision_number":2},{"id":"94d3443e41c50f7ac924f77ebe68b35bfc50610c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 17:12:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d01ac45e082f521286f23c1422605beda6aabd2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 18:20:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d7a1859bb324cef9129b5575bd23a03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1425a0720c4b4ab4a271e34be8d653e4 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5b14f0ab8f74a4b952e76635d35a313 : SUCCESS in 2m 58s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/21ecc25fec9b41d8b1f9a238df8e832d : SUCCESS in 49m 06s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/343af25c74dc4c00a8de689d10729966 : SUCCESS in 52m 15s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/70cdc92e46c5452683dba2e730d8cc7e : SUCCESS in 1h 00m 29s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/580366ec787c4f8688ff0d9c74219131 : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":2},{"id":"92cdf563e1e46958f94ac17cb91fa06f6e2bda36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 18:20:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d816f6501fe6961980f1531b1978743dfcd26f0b","revisions":{"8dd3e9fde8f36ef8bdf5e3f16e0c7d1bdca41056":{"kind":"REWORK","_number":1,"created":"2024-10-09 13:30:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/931918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/18/931918/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/18/931918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/18/931918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/18/931918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/18/931918/1"}}},"commit":{"parents":[{"commit":"2dfe0d24631ff10812fc70bcb35f726f19c2409d","subject":"Merge \"Consistently configure authtoken options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2dfe0d24631ff10812fc70bcb35f726f19c2409d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 13:29:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 13:29:25.000000000","tz":540},"subject":"Adopt to latest Puppet OpenStack modules","message":"Adopt to latest Puppet OpenStack modules\n\n* puppet-cinder\n * properties property of cinder_type now requires a Hash\n * support for VMWare backend has been removed\n* puppet-glance\n * properties property of glance_image now requires a Hash\n\nChange-Id: I028178976b96199cd773022f5b8fec70cf223bde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8dd3e9fde8f36ef8bdf5e3f16e0c7d1bdca41056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8dd3e9fde8f36ef8bdf5e3f16e0c7d1bdca41056"}]},"branch":"refs/heads/master"},"d816f6501fe6961980f1531b1978743dfcd26f0b":{"kind":"REWORK","_number":2,"created":"2024-10-09 13:32:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/931918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/18/931918/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/18/931918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/18/931918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/18/931918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/18/931918/2"}}},"commit":{"parents":[{"commit":"2dfe0d24631ff10812fc70bcb35f726f19c2409d","subject":"Merge \"Consistently configure authtoken options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2dfe0d24631ff10812fc70bcb35f726f19c2409d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 13:29:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 13:32:02.000000000","tz":540},"subject":"Adopt to latest Puppet OpenStack modules","message":"Adopt to latest Puppet OpenStack modules\n\n* puppet-cinder\n * properties property of cinder_type now requires a Hash\n * support for VMWare backend has been removed\n* puppet-glance\n * properties property of glance_image now requires a Hash\n\nChange-Id: I028178976b96199cd773022f5b8fec70cf223bde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d816f6501fe6961980f1531b1978743dfcd26f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d816f6501fe6961980f1531b1978743dfcd26f0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
