)]}'
{"id":"x%2Fpackstack~931954","triplet_id":"x%2Fpackstack~master~I498dfd013f2bf35297212de669d2a6ceb5cbdb63","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-15 08:55:51.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-11-15 08:55:51.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-11-15 07:42:37.000000000","reason":"\u003cGERRIT_ACCOUNT_31068\u003e replied on the change","reason_account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}},"hashtags":[],"change_id":"I498dfd013f2bf35297212de669d2a6ceb5cbdb63","subject":"Create volume types by built-in logic","status":"MERGED","created":"2024-10-09 14:27:49.000000000","updated":"2024-11-15 08:55:51.000000000","submitted":"2024-11-15 08:55:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931954","meta_rev_id":"9368bae9d8fe863de553a537308b183a54697e09","_number":931954,"virtual_id_number":931954,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 08:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-11-15 07:42:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-11-15 07:42:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 15:33:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-07 09:39:27.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-11-13 10:51:56.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"6cc611df35804762a21d41863f5f4fe7ac28c9a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 14:27:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f1dd1bbc4ad08062bba69f915392bdd16299131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 15:33:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f1e6b8cd3154a9791ed35d9ada50331\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d2b43f8778f4b518323fb08cc0655dd : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07ca83b97e734d4c8a0d916ac5e6e0dd : SUCCESS in 3m 14s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/929f01dfa8a64fc8b216604da6315b81 : FAILURE in 40m 28s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/71c968a66a9c4af7afe9529e144b45cc : SUCCESS in 50m 41s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/369736cfc1af4eb7bbff312d7018b21c : FAILURE in 57m 52s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e3ee4bd48d364f038fa0089d20edfc79 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7cc3d82d78371dcf0ca37da8eb3fb7c78b98f854","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-03 14:32:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c70417a0fb96612a6146050e7de74138e39a720c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-03 15:35:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3675737c698a43548664773f1f39304a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b6c391b8ee4352a286bd98e1cf8c88 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf9906a5b9b64f13b1ee90dd764fe9ab : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e65fe7282314551867c8c35de1bdde2 : SUCCESS in 3m 03s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2f20972ba50240cd98b8d4421b08b2f6 : SUCCESS in 40m 39s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/dd0376cbc8ea464f8627d568474825e0 : SUCCESS in 38m 21s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/585f32282dce4fc4b6f101183df3a182 : SUCCESS in 42m 30s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8b386a63f771435b9dcff55bd4f20989 : SUCCESS in 33m 39s","accounts_in_message":[],"_revision_number":2},{"id":"f7bead176d5dc882b00896aba8de90de09822793","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-11-07 09:39:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5b78afd1cbaaedad41ab6c719dd76ca1d810c7ab","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-13 10:51:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7207c6d17d64198e4afc862af9023ef98bdd9d36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 11:51:02.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/x/packstack for 931954,2","accounts_in_message":[],"_revision_number":2},{"id":"0090d5fce8690f728b1f200e45e44a8d452b583c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-13 12:24:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6c851f71ffabd4c738e876a40d3013698b52704c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 13:16:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f824b8e13c12421fb4b600d246226f03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8cbf6f3579345d193375a83095023df : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/659612603479449795b82b90c680fd13 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2459197beda4fb1b960cde6da532b13 : SUCCESS in 3m 04s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/edfc124d86ad487c83b8fcd996dbf2f7 : SUCCESS in 37m 39s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ea3cd3a177be487b96ad671707bec6ce : SUCCESS in 43m 23s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/fbb67627a12d4ffd9141da96e611f7f0 : SUCCESS in 46m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ef9be59d381645269fee46af20ff959c : SUCCESS in 48m 22s","accounts_in_message":[],"_revision_number":3},{"id":"9bbb5bec01541156adda4fc6b85bd6bf26eb0e05","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-14 16:26:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e2e29c4c7afc35496460bf84c93052cdaba094eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 16:27:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6eed0e46841281f99ac8d778a5efb182abd50ac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 17:25:35.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89f08d3df6e042809a9a3086e7a28b17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a7bd9d10484928aff06c72f20fca96 : SUCCESS in 2m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba917f6525c74e5384fcef77d7edb80b : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa52540eaaaf44e381afc2196550f6a0 : SUCCESS in 3m 32s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9b04f88b7e574a74857b2615889759a6 : SUCCESS in 34m 12s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ff0df7a48dc64a788520f58d2c9ce8ef : SUCCESS in 53m 17s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/85ccddb4ff404932bace1cd9b6052e05 : FAILURE in 25m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/be7a485f36914a69a2d61bb23b3f5380 : FAILURE in 11m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4e9b65919443f8ebcb78fc32a2ce1d6ac717e147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-14 17:57:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":4},{"id":"63dba7e0dddfe2a44572622afe21d3c5a0e0a92a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 19:03:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a6eba61df8246fbbe22639f5a00e829\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4a1398d4b7468986cce03e54dba3f3 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35081fd217c14d99b65dfdc4e257e4d4 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53673d7dc62f44179a7ee2a29292e0af : SUCCESS in 3m 18s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6dfb59f771cd4c9c856d5c833370de55 : SUCCESS in 53m 14s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1adb2a2533bd43d2ab3ea1975739e6d2 : SUCCESS in 57m 38s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6985aa6716bd4690b4405c40f4b3f26e : SUCCESS in 34m 27s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3c8f8c8035e94309b9d9807ddbf0e645 : SUCCESS in 44m 23s","accounts_in_message":[],"_revision_number":4},{"id":"8aed8bfc4e9e75a76b02e4daa562eb754a2eefce","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-15 07:42:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"27df063829fa155f15361423fffbee272034fd98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 07:43:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cca588fa0a438304ce937580e0a1d86edc745478","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 08:55:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecbc2f1837364e9ca01d40b09c6a1b62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e35c3068365f4898be9257f3de33dd33 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bc8d4912c6b403582159ce0d3260ab7 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9f72bd7402d42bd9081c86723d6c3dc : SUCCESS in 6m 25s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8348f1a73e724b0aab3a508df43bdc6f : SUCCESS in 41m 39s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0b83f2edea4e4b8194490e22760e6f4f : SUCCESS in 39m 13s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b673c85de5d94a9fbb0aa61bdc30c1b8 : SUCCESS in 1h 01m 40s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/09eebcb527504055bf34284166b71115 : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":4},{"id":"9368bae9d8fe863de553a537308b183a54697e09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 08:55:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"45745b6e220d75377f1fcc51dd062cd17203dbe3","revisions":{"6b611e9f545977ac6b807126f0b3b03ed8f93b62":{"kind":"REWORK","_number":1,"created":"2024-10-09 14:27:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/931954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/54/931954/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/54/931954/1"}}},"commit":{"parents":[{"commit":"d816f6501fe6961980f1531b1978743dfcd26f0b","subject":"Adopt to latest Puppet OpenStack modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d816f6501fe6961980f1531b1978743dfcd26f0b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 14:27:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 14:27:21.000000000","tz":540},"subject":"Create volume types by built-in logic","message":"Create volume types by built-in logic\n\nChange-Id: I498dfd013f2bf35297212de669d2a6ceb5cbdb63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6b611e9f545977ac6b807126f0b3b03ed8f93b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6b611e9f545977ac6b807126f0b3b03ed8f93b62"}]},"branch":"refs/heads/master"},"696f98ae40b56c5650bec9476287174c1b9b7121":{"kind":"REWORK","_number":2,"created":"2024-11-03 14:32:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/931954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/54/931954/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/54/931954/2"}}},"commit":{"parents":[{"commit":"d816f6501fe6961980f1531b1978743dfcd26f0b","subject":"Adopt to latest Puppet OpenStack modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d816f6501fe6961980f1531b1978743dfcd26f0b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 14:27:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-03 14:32:25.000000000","tz":540},"subject":"Create volume types by built-in logic","message":"Create volume types by built-in logic\n\nChange-Id: I498dfd013f2bf35297212de669d2a6ceb5cbdb63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/696f98ae40b56c5650bec9476287174c1b9b7121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/696f98ae40b56c5650bec9476287174c1b9b7121"}]},"branch":"refs/heads/master"},"b2bca8d009135ff57977caa05e18811299a90a48":{"kind":"REWORK","_number":3,"created":"2024-11-13 12:24:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/931954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/54/931954/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/54/931954/3"}}},"commit":{"parents":[{"commit":"472a7ca61c2102e503a4fb26dfc46496b3e98e20","subject":"Merge \"Drop deprecated glance::api::pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/472a7ca61c2102e503a4fb26dfc46496b3e98e20"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 14:27:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-13 12:23:41.000000000","tz":540},"subject":"Create volume types by built-in logic","message":"Create volume types by built-in logic\n\nChange-Id: I498dfd013f2bf35297212de669d2a6ceb5cbdb63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b2bca8d009135ff57977caa05e18811299a90a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b2bca8d009135ff57977caa05e18811299a90a48"}]},"branch":"refs/heads/master"},"45745b6e220d75377f1fcc51dd062cd17203dbe3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-14 17:57:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/54/931954/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/54/931954/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/54/931954/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/54/931954/4"}}},"commit":{"parents":[{"commit":"5d1c8ee31c564bbf032948cabb0ce92a512e1bf2","subject":"Merge \"Manila: Fix missing ssl options for RabbitMQ connection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5d1c8ee31c564bbf032948cabb0ce92a512e1bf2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 14:27:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-14 17:57:05.000000000","tz":0},"subject":"Create volume types by built-in logic","message":"Create volume types by built-in logic\n\nChange-Id: I498dfd013f2bf35297212de669d2a6ceb5cbdb63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/45745b6e220d75377f1fcc51dd062cd17203dbe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/45745b6e220d75377f1fcc51dd062cd17203dbe3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
