)]}'
{"id":"x%2Fpackstack~933178","triplet_id":"x%2Fpackstack~stable%2Frocky~I5136baf030916970cdd47f4e583b605785994c08","project":"x/packstack","branch":"stable/rocky","topic":"packstack-constraints","attention_set":{},"removed_from_attention_set":{"37373":{"account":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"last_update":"2024-10-24 07:02:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5136baf030916970cdd47f4e583b605785994c08","subject":"Updated from Packstack modules constraints","status":"ABANDONED","created":"2024-10-24 00:05:10.000000000","updated":"2024-10-24 07:02:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"42e0a8716f5275c94525e53813e01d413860f063","_number":933178,"virtual_id_number":933178,"owner":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-24 00:08:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 00:08:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f71bb31fff2c724dbe47f987c590f92e10c8bebf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2024-10-24 00:05:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07ea497300bb454ebed49fddda9c54a3e01ff1e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 00:08:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ce558f1eb674b16abe7b555f3510d24\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ea97a52cc247168abc266bc8a42e3a : FAILURE in 1m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/033f2e009fb5481c80cb6f8e678b4381 : FAILURE in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"42e0a8716f5275c94525e53813e01d413860f063","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-24 07:02:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de99c9d63dba1975076bb245c25bdddb81caf70d","revisions":{"de99c9d63dba1975076bb245c25bdddb81caf70d":{"kind":"REWORK","_number":1,"created":"2024-10-24 00:05:10.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/78/933178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/78/933178/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/78/933178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/78/933178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/78/933178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/78/933178/1"}}},"commit":{"parents":[{"commit":"3ee7479b85a5fa1ca93ef6330f1fcea3de687f45","subject":"Fix CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3ee7479b85a5fa1ca93ef6330f1fcea3de687f45"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2024-10-24 00:05:09.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2024-10-24 00:05:09.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I5136baf030916970cdd47f4e583b605785994c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/de99c9d63dba1975076bb245c25bdddb81caf70d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/de99c9d63dba1975076bb245c25bdddb81caf70d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
