)]}'
{"id":"x%2Fpackstack~933181","triplet_id":"x%2Fpackstack~stable%2Fwallaby~Ifcb71f107707cce976de6c6577dcaecdaa733a95","project":"x/packstack","branch":"stable/wallaby","topic":"packstack-constraints","attention_set":{},"removed_from_attention_set":{"37373":{"account":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"last_update":"2024-10-24 07:02:31.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifcb71f107707cce976de6c6577dcaecdaa733a95","subject":"Updated from Packstack modules constraints","status":"ABANDONED","created":"2024-10-24 00:08:22.000000000","updated":"2024-10-24 07:02:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"afe51760e07f215dacbffb3f6d6017189f05722f","_number":933181,"virtual_id_number":933181,"owner":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-24 00:20:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 00:20:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"384e90d44850bcebf5c7a5f90987f6f24159183e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2024-10-24 00:08:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cecc45e4faebc8c8c7fca387b5304666f30696e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 00:20:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc26abde47bd4c97b6612a2a92ee440e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37845cbaae2147aaaf1c2da41d4cb758 : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e79f095e8454a9e877abe64f18e6819 : FAILURE in 2m 00s","accounts_in_message":[],"_revision_number":1},{"id":"afe51760e07f215dacbffb3f6d6017189f05722f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-24 07:02:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9aab8ea107f8605a18c8650eee6070c9f585bf7d","revisions":{"9aab8ea107f8605a18c8650eee6070c9f585bf7d":{"kind":"REWORK","_number":1,"created":"2024-10-24 00:08:22.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/81/933181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/81/933181/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/81/933181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/81/933181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/81/933181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/81/933181/1"}}},"commit":{"parents":[{"commit":"0a42c916b23c825c790154a8dab02454ebc010bc","subject":"Explicitely disable ssl in mariadb server config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0a42c916b23c825c790154a8dab02454ebc010bc"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2024-10-24 00:08:22.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2024-10-24 00:08:22.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: Ifcb71f107707cce976de6c6577dcaecdaa733a95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9aab8ea107f8605a18c8650eee6070c9f585bf7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9aab8ea107f8605a18c8650eee6070c9f585bf7d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
