)]}'
{"id":"x%2Fpackstack~933435","triplet_id":"x%2Fpackstack~master~I1d941d8ea628055308f5ef3fc5c051cb902b54d4","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-31 07:28:43.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I1d941d8ea628055308f5ef3fc5c051cb902b54d4","subject":"Drop redundant nova resource dependencies","status":"MERGED","created":"2024-10-25 15:00:49.000000000","updated":"2024-10-31 09:25:16.000000000","submitted":"2024-10-31 09:25:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933435","meta_rev_id":"0430582c1e1cf10d9a945dd8802b0ef91b479f89","_number":933435,"virtual_id_number":933435,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-31 09:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2024-10-28 07:19:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-10-30 07:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-10-30 07:01:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-25 16:04:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-28 07:19:00.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-30 07:01:33.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"b9d99c28824d48c3f134787de6882966ed2d6903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-25 15:00:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e9625c2f3b85395c464074d8088623413b20115","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 16:04:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/978224ac9f794568a16251ca9bdd0397\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cef9c18c7824ee39b6489961b99ddcc : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31feecd3dec445118cf8f89792fbe997 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4d9d2c5d378442693eff7840a790771 : SUCCESS in 3m 41s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a6a489072683449cbe1590f08faa2749 : SUCCESS in 53m 04s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/621a72c7418b47e89bb56fea2d33ae73 : SUCCESS in 46m 35s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ac7299a6a59d4fcda5308d7d971416ca : SUCCESS in 58m 40s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/763f0c0c414641519d93edf659ef78ad : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0ccf706dcc093b2da690324e93153decdb779279","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-28 07:19:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ae735095e4f96b7fb656e241837ada8ba1af6f65","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-30 07:01:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a363eb601498857c94a8e4d1147bba67f3b7ee09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 07:02:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71595dda0e7e4b97399007f146c5348f4b2febc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 08:15:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e54be8bfad70492ba267d1b58c5b7c90\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95e726e90c4c4986bcd22d234935b5f2 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31779b304c914b90baa0ef17b5f9bbc9 : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d32f2cfc2055419880e421937ac8d907 : SUCCESS in 2m 30s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/89d5b633efa540bca8cd6438f769f2d0 : SUCCESS in 33m 30s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3852bd1a9c3f403c9ab9a0d431adcd3b : SUCCESS in 31m 34s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a63031fba3724f468d48db18839fa7be : SUCCESS in 1h 03m 58s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bb1bcc25628d49dfad029f29a8401004 : FAILURE in 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c112fecd10b1773543869d37ac95027733be4c2e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-30 12:18:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"815135f7a2351ff4d6da5c40ea064db27b0613ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 13:27:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd5d3a0ec3eb49f4b63c668cdd783816\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dedd46c624ab417781c0b5de700e4115 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01859daaf1314c02b0dc1c2c34541525 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d3caaa2de664281abf155a013ce4308 : SUCCESS in 3m 23s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d2d8ed0ca3c04e10a2a8b290795ddfff : SUCCESS in 36m 27s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/20945f444be44ceaa8c54ccefee3d200 : SUCCESS in 53m 01s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2554b9e9fe2f47c19f0ac4788d0ae7a8 : SUCCESS in 37m 30s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/60e2e67ce0f745f0ad598382268af104 : POST_FAILURE in 1h 02m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f6f650d8db26c2db6d0d6d512695a2af5b952dcd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-31 07:28:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54298b37213d7b487c0d70675074ac91fb329759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 08:18:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18689363193a4899b576596be635b6bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d981dcbdcb94a459b9f104b1360bc9d : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71274c0bc852418b8971f689bf62418a : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3931f20aec62424db1d6010f358f46cd : SUCCESS in 3m 54s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/3b63ae82f09748ac9a7085976985f868 : SUCCESS in 34m 02s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/daae1edb9fd74ec589178f0a667e71c7 : SUCCESS in 37m 57s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ca9c6a0dfb50492c8044c853617bfa5a : SUCCESS in 42m 04s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cd468042d6ec4609926ebcfffb15bc74 : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9d67bb2c08632769c427c7870766573d16008cfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 08:19:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3af2542d4599885e9481831dea5a7a9f8becf606","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 09:25:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d65adb401844114822264ee8823ae61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82a06c9bc15a45f5895ae25b44070b25 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57d0ed6f9636411a816c076e334db6ad : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b8526f794e04a30a629b318de2f6975 : SUCCESS in 6m 09s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0f92f5d517914c8bb3a905b02a221de3 : SUCCESS in 33m 20s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f816547eb5724f1e9a1f2b3b56b7e9c4 : SUCCESS in 40m 39s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d589891720394b4896c4bebef86fffae : SUCCESS in 38m 37s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c35804844040412e995e6aa54dacd96a : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0430582c1e1cf10d9a945dd8802b0ef91b479f89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 09:25:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9111f2adfc5ab63c1a4529cf3d1a13092c6407ad","revisions":{"9111f2adfc5ab63c1a4529cf3d1a13092c6407ad":{"kind":"REWORK","_number":1,"created":"2024-10-25 15:00:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/933435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/35/933435/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/35/933435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/35/933435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/35/933435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/35/933435/1"}}},"commit":{"parents":[{"commit":"a55614ca40958e4c2e99f4011b1907af73b3fe2c","subject":"Merge \"Add support of Python 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a55614ca40958e4c2e99f4011b1907af73b3fe2c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 14:59:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 15:00:35.000000000","tz":540},"subject":"Drop redundant nova resource dependencies","message":"Drop redundant nova resource dependencies\n\nCurrent nova_flavor resource automatically requires that all nova\nservices are up. Also nova user credential is always configured before\nnova services are all up.\n\nChange-Id: I1d941d8ea628055308f5ef3fc5c051cb902b54d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9111f2adfc5ab63c1a4529cf3d1a13092c6407ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9111f2adfc5ab63c1a4529cf3d1a13092c6407ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
