)]}'
{"id":"x%2Fpackstack~933440","triplet_id":"x%2Fpackstack~master~I9cfbd0c84037bfe0b532d8bacdc77b55e631efd9","project":"x/packstack","branch":"master","topic":"api-mod_wsgi","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-13 11:49:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9cfbd0c84037bfe0b532d8bacdc77b55e631efd9","subject":"Cinder: Switch to httpd + mod_wsgi","status":"MERGED","created":"2024-10-25 16:19:48.000000000","updated":"2024-11-13 11:49:25.000000000","submitted":"2024-11-13 11:49:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933440-api-mod_wsgi","meta_rev_id":"76f73e061a1e94341904ea7fcdce78edb1c156c8","_number":933440,"virtual_id_number":933440,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-13 11:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2024-11-03 10:52:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-11-13 10:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-11-13 10:46:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-25 19:04:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-03 10:52:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-11-13 10:46:32.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"9c45f6d237582fa081a748b2dca0dcf43d00fe54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-25 16:19:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b210bd837369de73e6c7ca6b2cb3c2bec29d83c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-25 17:36:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5666b72361c89a953b2efba1103a8fb9c07cbb74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 19:04:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c7544ce672d4b028e839fcd610d236c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5fff72d83e24c44a5151e9c523ac19b : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91857f884bcf47078283bd912cc969fb : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9649d07416574ff6988fb775ba9f6b63 : SUCCESS in 2m 46s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1573851db4fc447b89b80bef760375f0 : FAILURE in 1h 20m 40s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c66542a67d554289a888a72821530c84 : SUCCESS in 26m 43s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/13836eb45413490790a56686a27babb4 : FAILURE in 1h 22m 39s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a003c2bd5a8848b2ade83b0a6db38b03 : FAILURE in 35m 44s","accounts_in_message":[],"_revision_number":2},{"id":"039d9109a72413c13b6d57cf3004a39117a54a26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 02:28:49.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"14dda637ec927c25bba85b9e18245fb051d04b0f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 02:28:56.000000000","message":"Topic set to api-mod_wsgi","accounts_in_message":[],"_revision_number":3},{"id":"d8805596ecdb93b97d5fd2ac1d596d82a5e80156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 03:11:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7815c4fe3eb4187a40c63793c947a16\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3199304ae23743399e8ba37f2194a0ab : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be13fbd1d8b5407baa748ff9af38d444 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b84c6314505743ed967ec1b9c37b98e2 : SUCCESS in 5m 56s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/966676402c6c4264aad96e46785f5be8 : FAILURE in 21m 17s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8b4e402137434bb19de61091b86f44df : SUCCESS in 31m 28s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a3b1f0d266d44b2d8fda92ce3bda59a5 : FAILURE in 27m 15s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/82fcc1f2e8704a40bf6914e2d94f06bc : FAILURE in 25m 41s","accounts_in_message":[],"_revision_number":3},{"id":"eb5790563e7f92187659b16f12c81d991bfa9072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-03 09:31:02.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e499ff941720fbfddbd7e3081e10e676b560d039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-03 10:31:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb188514c21d44a1a4626c06bd1e9d13\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ea5d6ed4e344bc1b42df1f7b182118e : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41b806b135b44cb39002962eb63c000a : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c89fb07b0c04f7fa593d887355fc48d : SUCCESS in 3m 43s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/60923dcef90943dcb89e36fbca3c6da6 : SUCCESS in 29m 40s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1ddd5219129f410d921abd10f93320bb : SUCCESS in 39m 20s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a808158a805c49fd89922b495c0b84da : SUCCESS in 55m 22s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6aaa588c84714bee988b4cc0c36017c8 : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":4},{"id":"5220bcdc211de218baf19b05299a9468e5c8069d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-11-03 10:52:36.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5a7cd7ea86ff98e07a7c431dafdc330fe2f90805","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-13 10:46:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"53e31b03a607df6aab09ce02cc870085f3a7c4d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 10:47:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fb76a8edb706e555079255e5288917c39fd5fe07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 11:49:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a72ace9574249f09a58514fda8da7f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/016895e240504f8498090ed1b154be34 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a1d75a2f3234d278c37a6f68accc9fc : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8e40cf9f7b04c1fa271699510ac6486 : SUCCESS in 3m 21s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6b2b9466090d4fa4b01aa511763a53eb : SUCCESS in 54m 45s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/53c9c1f213a047fbb80d49dc8d9906b1 : SUCCESS in 51m 04s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/aec440dc7f4d41bfa681b18ab9cca8f5 : SUCCESS in 29m 13s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/035c53c029cd4e01bbb4119066e27825 : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":4},{"id":"76f73e061a1e94341904ea7fcdce78edb1c156c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 11:49:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e43c335113511d93abf3a6a20a37067c34b23677","revisions":{"261480fe6ee2ce1a2f5bea5f8efb36ee80a44c95":{"kind":"REWORK","_number":1,"created":"2024-10-25 16:19:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/933440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/40/933440/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/40/933440/1"}}},"commit":{"parents":[{"commit":"a55614ca40958e4c2e99f4011b1907af73b3fe2c","subject":"Merge \"Add support of Python 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a55614ca40958e4c2e99f4011b1907af73b3fe2c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 16:19:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 16:19:37.000000000","tz":540},"subject":"Cinder: Switch to httpd + mod_wsgi","message":"Cinder: Switch to httpd + mod_wsgi\n\nChange-Id: I9cfbd0c84037bfe0b532d8bacdc77b55e631efd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/261480fe6ee2ce1a2f5bea5f8efb36ee80a44c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/261480fe6ee2ce1a2f5bea5f8efb36ee80a44c95"}]},"branch":"refs/heads/master"},"85a4425181ee0cf1620c6fb66525323d26b903b1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-25 17:36:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/933440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/40/933440/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/40/933440/2"}}},"commit":{"parents":[{"commit":"abd01d8be96ffc295773a24cf165b80dbf75323a","subject":"Drop redundant cinder resource dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/abd01d8be96ffc295773a24cf165b80dbf75323a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 16:19:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 17:36:16.000000000","tz":0},"subject":"Cinder: Switch to httpd + mod_wsgi","message":"Cinder: Switch to httpd + mod_wsgi\n\nChange-Id: I9cfbd0c84037bfe0b532d8bacdc77b55e631efd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/85a4425181ee0cf1620c6fb66525323d26b903b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/85a4425181ee0cf1620c6fb66525323d26b903b1"}]},"branch":"refs/heads/master","description":"Rebase"},"86d93a20f2fb15172ccebb6db50e5375e7702bae":{"kind":"REWORK","_number":3,"created":"2024-10-26 02:28:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/933440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/40/933440/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/40/933440/3"}}},"commit":{"parents":[{"commit":"abd01d8be96ffc295773a24cf165b80dbf75323a","subject":"Drop redundant cinder resource dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/abd01d8be96ffc295773a24cf165b80dbf75323a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 16:19:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 02:28:49.000000000","tz":0},"subject":"Cinder: Switch to httpd + mod_wsgi","message":"Cinder: Switch to httpd + mod_wsgi\n\nChange-Id: I9cfbd0c84037bfe0b532d8bacdc77b55e631efd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/86d93a20f2fb15172ccebb6db50e5375e7702bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/86d93a20f2fb15172ccebb6db50e5375e7702bae"}]},"branch":"refs/heads/master"},"e43c335113511d93abf3a6a20a37067c34b23677":{"kind":"REWORK","_number":4,"created":"2024-11-03 09:31:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/933440/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/40/933440/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/40/933440/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/40/933440/4"}}},"commit":{"parents":[{"commit":"abd01d8be96ffc295773a24cf165b80dbf75323a","subject":"Drop redundant cinder resource dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/abd01d8be96ffc295773a24cf165b80dbf75323a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 16:19:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-03 09:31:02.000000000","tz":0},"subject":"Cinder: Switch to httpd + mod_wsgi","message":"Cinder: Switch to httpd + mod_wsgi\n\nChange-Id: I9cfbd0c84037bfe0b532d8bacdc77b55e631efd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e43c335113511d93abf3a6a20a37067c34b23677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e43c335113511d93abf3a6a20a37067c34b23677"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
