)]}'
{"id":"x%2Fpackstack~933477","triplet_id":"x%2Fpackstack~master~Idd08721acdb294b9a24f6e43aec1acfaa5f2d27d","project":"x/packstack","branch":"master","topic":"packstack-constraints","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-10-30 08:15:12.000000000","reason":"Change was submitted"},"37373":{"account":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"last_update":"2024-10-30 08:15:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd08721acdb294b9a24f6e43aec1acfaa5f2d27d","subject":"Updated from Packstack modules constraints","status":"MERGED","created":"2024-10-26 00:03:00.000000000","updated":"2024-10-30 08:15:12.000000000","submitted":"2024-10-30 08:15:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933477-packstack-constraints","meta_rev_id":"66bc757f13995ad8b8e717a407706192200a7032","_number":933477,"virtual_id_number":933477,"owner":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-30 08:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-30 06:57:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-30 06:57:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-26 01:02:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-28 07:17:30.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-30 06:57:21.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"42630cd05ef1b889a511ad68fbc6e8d3adc9811f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2024-10-26 00:03:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b536b772aa2e33f7dc43eee28549a2a6004dd44e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 01:02:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94aa36bfbd8f49c8af7152bec4d80834\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9906842e9dfc4a479297ea5e8cab4727 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41c0a6b5c27b4776ade5196b967bf1dc : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/182fd4c9a4e348aa99111f3ed04ef4de : SUCCESS in 3m 46s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b853f5af415f44cbb9d9b5a41574406a : SUCCESS in 54m 19s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3328b4d2c83c469fa5280dd399ea880c : SUCCESS in 31m 09s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/973dacf3aa794326866fd6ff8ecc6d8c : SUCCESS in 54m 50s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/330bf670fe7d4ada8726810839069f91 : SUCCESS in 35m 22s","accounts_in_message":[],"_revision_number":1},{"id":"59f96edd180f5c83933768e664843da4c865d63f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-28 07:17:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2080e958e135cf54cfbbf933b572fecea11ead06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2024-10-30 00:03:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"23ca0917ede2cdd81d6c026bda535121df5cedee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 01:09:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c54e77522b1481f9024d43b58275646\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75cd8f0377842f1a29feb123814c2ee : SUCCESS in 2m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd14a4464a2d47aba19327a5ebe53e41 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33d98e8d3e794f4490acdd8beabc3601 : SUCCESS in 2m 59s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/04e8483109464c3fa23ce9aac09ab861 : SUCCESS in 53m 56s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6096e31337c4464a9164de8ebccbe13d : SUCCESS in 53m 44s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ab7ed9559429483b9c0cb372f36d8504 : SUCCESS in 31m 25s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d7ee36db9dff4367927c19444e5212ab : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b0a317a4cc8245ecd7d59b5bd882bb10f31adb86","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-30 06:57:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c61dd68e373b3ab0e7c1ae8965290fe06a4ceb78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 06:57:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2045d358662f531d68e470c5eba671606f87a029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 08:15:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0859a339b3754323a806e96e5dbe498b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b9c6ea47b44829a6472118378cf50b : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4aabdfc26b5e40f4aa50e9975dfbb883 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b7acaae8b274973acb4a158060b3ac3 : SUCCESS in 3m 07s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/58bf8fb72279436593b14b7469be3a6b : SUCCESS in 34m 22s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/a3adfe719d63475ea31f745103bad097 : SUCCESS in 29m 10s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/dfca885c1b7d4dc3918e94ebbf24953f : SUCCESS in 1h 11m 48s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/de4135c3b6d649faab773f4e84973403 : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":2},{"id":"66bc757f13995ad8b8e717a407706192200a7032","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 08:15:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f0b19989378d06943e2b9835dc52ce084e536499","revisions":{"a5626670176d5a1ea67b88a630ca199e848ee8b2":{"kind":"REWORK","_number":1,"created":"2024-10-26 00:03:00.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/77/933477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/77/933477/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/77/933477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/77/933477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/77/933477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/77/933477/1"}}},"commit":{"parents":[{"commit":"a55614ca40958e4c2e99f4011b1907af73b3fe2c","subject":"Merge \"Add support of Python 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a55614ca40958e4c2e99f4011b1907af73b3fe2c"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2024-10-26 00:02:58.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2024-10-26 00:02:58.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: Idd08721acdb294b9a24f6e43aec1acfaa5f2d27d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a5626670176d5a1ea67b88a630ca199e848ee8b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a5626670176d5a1ea67b88a630ca199e848ee8b2"}]},"branch":"refs/heads/master"},"f0b19989378d06943e2b9835dc52ce084e536499":{"kind":"REWORK","_number":2,"created":"2024-10-30 00:03:07.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/77/933477/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/77/933477/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/77/933477/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/77/933477/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/77/933477/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/77/933477/2"}}},"commit":{"parents":[{"commit":"a55614ca40958e4c2e99f4011b1907af73b3fe2c","subject":"Merge \"Add support of Python 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a55614ca40958e4c2e99f4011b1907af73b3fe2c"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2024-10-26 00:02:58.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2024-10-30 00:03:06.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: Idd08721acdb294b9a24f6e43aec1acfaa5f2d27d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f0b19989378d06943e2b9835dc52ce084e536499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f0b19989378d06943e2b9835dc52ce084e536499"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
