)]}'
{"id":"x%2Fpackstack~933480","triplet_id":"x%2Fpackstack~master~I7da0f1fedd3ccbbf1e6cdadc9edb0a4c17aae36d","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-13 11:51:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7da0f1fedd3ccbbf1e6cdadc9edb0a4c17aae36d","subject":"Drop deprecated glance::api::pipeline","status":"MERGED","created":"2024-10-26 02:36:23.000000000","updated":"2024-11-13 11:51:01.000000000","submitted":"2024-11-13 11:51:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933480","meta_rev_id":"2ba33eef2b86157807abab1def4904bd2dd5ebfb","_number":933480,"virtual_id_number":933480,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-13 11:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2024-11-01 08:38:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-13 10:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-13 10:49:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-26 03:42:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-01 08:38:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-11-13 10:49:35.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"1d902e8a99d4304b67675fb15e544537c8fcfb11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 02:36:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c087e3e4c8c7daf0d028bf2837f66f9142e1bc4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 03:42:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b2639819a8541858d163ba14753e806\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada3436dad2f47058e53945641512a7a : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/769227d281d34f8d9058fac89b13e8c0 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc99e88091344ba8990f5bb1d6838b4f : SUCCESS in 4m 36s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/281b098f7bd5459bb355539cbbdb3dff : SUCCESS in 40m 16s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/91294a422b984b33bbdd92c880361735 : SUCCESS in 52m 12s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0d2767cffc1a47409fc2b14ba06fbbc8 : FAILURE in 14m 06s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2b972bda918e4b5b830f8760a12c8c76 : FAILURE in 35m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2c91a3200c6de0a459a2d87a500df3339937313d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-01 06:27:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"71c0922a488cc7d2c42b08fd78070ad23de4c7e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 07:27:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b1c069a175143e3be9241b459d36bbb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0be8dcb7a8e4febb00837d296a2562e : SUCCESS in 2m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/501d29a3e05749a9a161d338c54fe2b6 : SUCCESS in 2m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a92d5fd32f6646dbb2b7a49b6dbb3876 : SUCCESS in 3m 04s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/aceafbdab746434ba694fce80836e47e : SUCCESS in 28m 08s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/029f82b4192a462fbcb41e48733838fc : SUCCESS in 39m 48s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8b9cfddb46784f7986f81233a76abdd1 : SUCCESS in 38m 04s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/be0a4cf43c7f4e6d9f54c64385defb10 : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1ca0d20219dfee59e84ef60ed0ccebdc6b7906b6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-11-01 08:38:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"80019f60a16af6935cfdba822a84a6c2da5f3d7d","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-13 10:49:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"60aba9e9b84d465633ce6227ef937063806b4d37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 10:50:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc07b30d354a84ea5970ff8233dce1f6d00a2d81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 11:51:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e63b9c5abe8e4fc093aad9fcbedc3e3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a59ae68f96b43a9be1f5af08534c251 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1607f558648148809133081afb429e02 : SUCCESS in 5m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51aa2af758bf417498f164f1002d7179 : SUCCESS in 3m 45s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6039be34f31f46d688fd95b645f2361c : SUCCESS in 41m 58s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/a5274eb97ea54fc1b62b760d6677a922 : SUCCESS in 27m 59s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/fbfcb058624a4b10867f7df4e506d1f5 : SUCCESS in 53m 10s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/40be58f609384bdcb02de6499f311489 : SUCCESS in 47m 32s","accounts_in_message":[],"_revision_number":2},{"id":"2ba33eef2b86157807abab1def4904bd2dd5ebfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 11:51:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0672a9742354e7afc814fac6a17dbb4f22865699","revisions":{"737bfcdc8c9fc8202f4ea3da88ed9f6eacefae25":{"kind":"REWORK","_number":1,"created":"2024-10-26 02:36:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/933480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/933480/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/933480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/933480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/933480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/933480/1"}}},"commit":{"parents":[{"commit":"a55614ca40958e4c2e99f4011b1907af73b3fe2c","subject":"Merge \"Add support of Python 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a55614ca40958e4c2e99f4011b1907af73b3fe2c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 02:35:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 02:35:38.000000000","tz":540},"subject":"Drop deprecated glance::api::pipeline","message":"Drop deprecated glance::api::pipeline\n\nThis parameter was deprecated a few cycles ago[1].\n\n[1] fa5fcb4ab5f861003be7409fdcf71c94120d0b5e\n\nChange-Id: I7da0f1fedd3ccbbf1e6cdadc9edb0a4c17aae36d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/737bfcdc8c9fc8202f4ea3da88ed9f6eacefae25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/737bfcdc8c9fc8202f4ea3da88ed9f6eacefae25"}]},"branch":"refs/heads/master"},"0672a9742354e7afc814fac6a17dbb4f22865699":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-01 06:27:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/933480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/933480/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/933480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/933480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/933480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/933480/2"}}},"commit":{"parents":[{"commit":"f3de962c3741204bae6c7738dcc6613ad04fd81e","subject":"Merge \"Drop redundant nova resource dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f3de962c3741204bae6c7738dcc6613ad04fd81e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 02:35:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-01 06:27:02.000000000","tz":0},"subject":"Drop deprecated glance::api::pipeline","message":"Drop deprecated glance::api::pipeline\n\nThis parameter was deprecated a few cycles ago[1].\n\n[1] fa5fcb4ab5f861003be7409fdcf71c94120d0b5e\n\nChange-Id: I7da0f1fedd3ccbbf1e6cdadc9edb0a4c17aae36d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0672a9742354e7afc814fac6a17dbb4f22865699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0672a9742354e7afc814fac6a17dbb4f22865699"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
