)]}'
{"id":"x%2Fpackstack~933493","triplet_id":"x%2Fpackstack~master~I22b425a8380742b163f34bccfb356755e7339602","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-14 15:55:36.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-11-14 15:55:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I22b425a8380742b163f34bccfb356755e7339602","subject":"Use helper functions to compose url strings","status":"MERGED","created":"2024-10-26 14:07:05.000000000","updated":"2024-11-14 15:55:36.000000000","submitted":"2024-11-14 15:55:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"933493","meta_rev_id":"c46f3832c738e3298897630cef706ecfabecdaa4","_number":933493,"virtual_id_number":933493,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 15:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2024-10-28 07:16:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-11-14 14:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-11-14 14:43:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-26 14:55:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-28 07:16:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-11-14 14:43:19.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"ef0886328ca25a4b2b87f07404a9ea6cb8a05f80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 14:07:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb89ecd64d9058cb7a3df887f42657f6e47f0061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 14:11:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"211d9a8c915e0c61fabf324afa5918f5f3b23dd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 14:55:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a365c2bbcad434f800f43ebafde58fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531a8d6d516a428a94eb4bbcf77f6e5b : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d8f83ddacd848deb220e2c7347ba200 : SUCCESS in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd565f2c4ad843e5a00d4ded719a719a : SUCCESS in 3m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5b2193c1dc104d3b81e19d9e913c41f6 : SUCCESS in 38m 45s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f2e61b59463d474cb2cc1ab4ca15ecc1 : SUCCESS in 33m 38s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1561dd3c8cdf4b9aba962fdb4f000672 : SUCCESS in 35m 39s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/919b732f7b7147bebf31fdbc79757c9e : SUCCESS in 35m 34s","accounts_in_message":[],"_revision_number":2},{"id":"efe25e33720a12ff8c0cd304f845ae2f5b14a198","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-28 07:16:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb719380909e79f39d0dd5976d6e5a31dc86f9b1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-28 13:53:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ebefbbb559dac3e3c50a4bf920d844c18f790e7","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-14 14:43:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"47b483e08bd38dadf9f557d2b5d8be4303d96a2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 14:43:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"94efa65a5882dc18c16f4c30f6fca97b55ff453a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 15:55:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d6bfc7182574c41aa4c751c00acd93a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b273ef00b2049068916f248523d68a1 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffb7f4b936be46cf8fb54479f2bbfe42 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee2ac076af4f4317bcfe7605e87f9213 : SUCCESS in 3m 30s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/41c61cb7b51f41ddab3add427648362f : SUCCESS in 55m 11s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8a91fa02a2b34df3ad559758a3cfeb5b : SUCCESS in 39m 23s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/42fb296e716840c89a1e50ed920a82ac : SUCCESS in 46m 35s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9c009c4b80e9402da6765d6c5a5f3415 : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c46f3832c738e3298897630cef706ecfabecdaa4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 15:55:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b31ffa1d0a149372bc12d88ceefe123cf21a1b99","revisions":{"2c5f566af82041c5c4d2e9763a12c57984208c30":{"kind":"REWORK","_number":1,"created":"2024-10-26 14:07:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/933493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/93/933493/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/93/933493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/93/933493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/93/933493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/93/933493/1"}}},"commit":{"parents":[{"commit":"a55614ca40958e4c2e99f4011b1907af73b3fe2c","subject":"Merge \"Add support of Python 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a55614ca40958e4c2e99f4011b1907af73b3fe2c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 14:06:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 14:06:54.000000000","tz":540},"subject":"Use helper functions to compose url strings","message":"Use helper functions to compose url strings\n\npuppet-openstacklib provides a few useful methods to build url strings\nfrom elements. Use these helper methods to simplify the logic.\n\nChange-Id: I22b425a8380742b163f34bccfb356755e7339602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2c5f566af82041c5c4d2e9763a12c57984208c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2c5f566af82041c5c4d2e9763a12c57984208c30"}]},"branch":"refs/heads/master"},"b31ffa1d0a149372bc12d88ceefe123cf21a1b99":{"kind":"REWORK","_number":2,"created":"2024-10-26 14:11:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/933493/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/93/933493/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/93/933493/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/93/933493/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/93/933493/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/93/933493/2"}}},"commit":{"parents":[{"commit":"a55614ca40958e4c2e99f4011b1907af73b3fe2c","subject":"Merge \"Add support of Python 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a55614ca40958e4c2e99f4011b1907af73b3fe2c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 14:06:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 14:09:53.000000000","tz":540},"subject":"Use helper functions to compose url strings","message":"Use helper functions to compose url strings\n\npuppet-openstacklib provides a few useful methods to build url strings\nfrom elements. Use these helper methods to simplify the logic.\n\nChange-Id: I22b425a8380742b163f34bccfb356755e7339602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b31ffa1d0a149372bc12d88ceefe123cf21a1b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b31ffa1d0a149372bc12d88ceefe123cf21a1b99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
