)]}'
{"id":"x%2Fpackstack~933566","triplet_id":"x%2Fpackstack~master~Ib72a25f2946c568f4a0ee2469d9e146abc1670e9","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-14 19:05:18.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-11-14 17:56:58.000000000","reason":"\u003cGERRIT_ACCOUNT_31068\u003e replied on the change","reason_account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}},"hashtags":[],"change_id":"Ib72a25f2946c568f4a0ee2469d9e146abc1670e9","subject":"Remove unnecessary x bit","status":"MERGED","created":"2024-10-28 13:18:54.000000000","updated":"2024-11-14 19:05:18.000000000","submitted":"2024-11-14 19:05:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933566","meta_rev_id":"9ce5249ff0553b7143460d9ee50669758385a456","_number":933566,"virtual_id_number":933566,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 19:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2024-11-14 16:26:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-11-14 17:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-11-14 17:56:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-28 14:36:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-29 09:33:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-11-14 14:43:03.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"e34ab068a0bea0bfc7062a338eb08d377e77e4be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-28 13:18:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7cc7c382efa2b5ee604affa6edf796ad510aae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 14:36:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/293289b83547461690a611e8e5ce37cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8589e29a9bca46f7a5d116c73bf7b13e : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c710c717490543a1a70f102dd449d157 : SUCCESS in 5m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe6cb892347d4855a38f742fe3f7fd44 : SUCCESS in 3m 36s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/43d9fb6e2ad542e9abbaf3377508d3d0 : SUCCESS in 37m 12s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c8d4635b95a349f99f2a00cbc45100aa : SUCCESS in 35m 01s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/24099859364f4bf393ca6547df3143a7 : SUCCESS in 45m 34s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/113e4f32db334c76be6050082305e889 : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":1},{"id":"597715c2564fddcdba7ad5fd8c1d9d7e58d04a5d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-29 09:33:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dbebf186ccb63be57eeb197fe95ec2f37a985920","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-14 14:43:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be49c21b6acb8c8df9ed2f2653a0f2e61886461d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 15:55:35.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/x/packstack for 933566,1","accounts_in_message":[],"_revision_number":1},{"id":"07feb84b7d100ee465625ffffcf98c2f9dd687d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-14 16:26:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"6f8dc6e976cec63baf9eb792439859b5aed67d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 17:26:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43283744cf654aeb872c284ede0af369\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b639f0fcaa1b499a917198e134820476 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/888b739d6a1f41b4b08ca3cfd18497f9 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4d45fd19aa84e92b3c1dd55d9b95ebc : SUCCESS in 3m 42s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e14e59ec36fb4d13bd01ade3383ed7ea : SUCCESS in 29m 28s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b62a0b24c05041cf8fe7af874936d96f : SUCCESS in 54m 37s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7118a80775484f78897894d20a359853 : SUCCESS in 37m 30s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c11300ecff464f0c889c0c6d3471b45d : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e2c083c5812c330d7963966f733dd14caa8543c6","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-14 17:56:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bb475e52ea5f8da256e46e7837a523695ebff47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 17:57:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b6d452dc86398ae89377b9a6790d5ba346640185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 18:55:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/913f1195d50e4e54bb31da020982da4d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e370e76ec8884aa3b7c6b1d5469abc1f : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b0f4184df3c44c383cbb364f56fe2f0 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bb620c40f824ef68fb96706c955ad72 : SUCCESS in 2m 26s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4e5086766fce4d92a548df33087d69b3 : SUCCESS in 53m 42s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d59c6d2b062043f6b390f5d25e8843ad : SUCCESS in 34m 16s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/785ecfa071ae401db7b1d1ee33ad5811 : SUCCESS in 40m 43s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d24544174cea4d1db6594c6663d73016 : SUCCESS in 40m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b2af10caab8459c2d695ade5fba07934edbee26d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 19:05:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3e5f6d8f4a14645b72b1307d47d8002\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/434277c4c3a8428995e724e917a2d1f5 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a4c4fcac750499c985cc2030f2ed43f : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f1a6335c24704e36ae751c819922cefe : SUCCESS in 5m 30s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/dc5b93358371452994934fb61b9aede6 : SUCCESS in 45m 25s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/a765fdf5a685417c98d049e91ab4ed6c : SUCCESS in 58m 32s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/925bd918f1ee43b9b8364acfd5715f4c : SUCCESS in 46m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3c432a7911b64fa38074728f9755f781 : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":2},{"id":"9ce5249ff0553b7143460d9ee50669758385a456","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 19:05:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e313c8fd7ccb0a2723dc74c6dae5ce13cfabf83","revisions":{"bf105049a391db4e2d9da4832294b5f4ed9bc990":{"kind":"REWORK","_number":1,"created":"2024-10-28 13:18:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/933566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/66/933566/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/66/933566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/66/933566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/66/933566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/66/933566/1"}}},"commit":{"parents":[{"commit":"a55614ca40958e4c2e99f4011b1907af73b3fe2c","subject":"Merge \"Add support of Python 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a55614ca40958e4c2e99f4011b1907af73b3fe2c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-28 13:18:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-28 13:18:21.000000000","tz":540},"subject":"Remove unnecessary x bit","message":"Remove unnecessary x bit\n\n... and coding header which is no longer required in Pythhon 3.\n\nChange-Id: Ib72a25f2946c568f4a0ee2469d9e146abc1670e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bf105049a391db4e2d9da4832294b5f4ed9bc990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bf105049a391db4e2d9da4832294b5f4ed9bc990"}]},"branch":"refs/heads/master"},"0e313c8fd7ccb0a2723dc74c6dae5ce13cfabf83":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-14 16:26:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/66/933566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/66/933566/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/66/933566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/66/933566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/66/933566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/66/933566/2"}}},"commit":{"parents":[{"commit":"5d1c8ee31c564bbf032948cabb0ce92a512e1bf2","subject":"Merge \"Manila: Fix missing ssl options for RabbitMQ connection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5d1c8ee31c564bbf032948cabb0ce92a512e1bf2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-28 13:18:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-14 16:26:17.000000000","tz":0},"subject":"Remove unnecessary x bit","message":"Remove unnecessary x bit\n\n... and coding header which is no longer required in Pythhon 3.\n\nChange-Id: Ib72a25f2946c568f4a0ee2469d9e146abc1670e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0e313c8fd7ccb0a2723dc74c6dae5ce13cfabf83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0e313c8fd7ccb0a2723dc74c6dae5ce13cfabf83"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
