)]}'
{"id":"x%2Fpackstack~935480","triplet_id":"x%2Fpackstack~master~Ibbcdb9a46bf02057a3640d8260f76a13f94e3882","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-17 10:01:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbcdb9a46bf02057a3640d8260f76a13f94e3882","subject":"Enable caching","status":"MERGED","created":"2024-11-18 06:56:37.000000000","updated":"2024-12-17 10:01:41.000000000","submitted":"2024-12-17 10:01:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935480","meta_rev_id":"cf4c7beb157560d9f546c81bfd5ab190837b4c7e","_number":935480,"virtual_id_number":935480,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-17 10:01:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":2,"date":"2024-12-17 08:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":1,"date":"2024-12-17 08:42:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-18 08:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-17 08:42:58.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"10d40ab31d4c813b4a7d472263deeb1faa4b1cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 06:56:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8854ed00f3d87c6e8c00bf031e65ad23fa8b2d3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 06:59:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65984870aef953057a6b8a117d8a724c0ad67d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 08:43:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a2a9fb2136e43fb9b6e7c176ce91111\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27fab8502da4a79b309af83f8ece73c : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/030bfdf066614663b2de423ef5a463a3 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/722c7f07be5c4c67a338f9bbf81077ed : SUCCESS in 3m 06s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/134113adfc77464db61a6d26c1932ad0 : SUCCESS in 56m 56s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b9a1b433d4324d8e96d8b3128b617cf6 : SUCCESS in 28m 33s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b8e90167ecc749eab2868a35faf0b935 : FAILURE in 1h 37m 28s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b6b2db9e38224c81926f10d7c69d4549 : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d8568b5ce9edca0ceba4188846f39f3cbd0cf138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 14:25:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0ea7ca9baceae8b0b3225c306ef0a7a1f35d8f2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 14:57:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9c02fa6fd674f278dd19c4d3ec6e924\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcbd791644e54956b83b65e77d138cf0 : SUCCESS in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c823d1972eaf4cfa824a82d3ecd104d4 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7947b0848302472b903862a4d1f1e386 : SUCCESS in 3m 07s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4128a326f0e748209219ce41e4d4ae8b : FAILURE in 10m 05s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7680ac73f1be4009b5daf5783ed46219 : FAILURE in 11m 58s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/4bba5eb24a084e73bbb81f07aeaf8b84 : FAILURE in 22m 35s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cd3c9cfe461846a485115bc144040176 : FAILURE in 26m 24s","accounts_in_message":[],"_revision_number":3},{"id":"1c50a333f376d2d12cb9c3e12f641ae447ad1617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 15:02:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"904158b23835d07d1c1a076615aee7cc7c43957e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 16:15:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c71a04b6e9cb4ed3abc30ebd8095edaa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43067f22cae04af18c0aa8ad53c3d2d0 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d07b3f5807be494e980401ce6f80fc72 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ad42a2f30654056bc930b16d53e9ae6 : SUCCESS in 2m 38s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c5e34803dea544aa832c0b3666d98698 : SUCCESS in 29m 36s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b8b9d1d9ad0f42f188ac6e9d97ba753f : SUCCESS in 38m 06s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/fe3ef63fa8374a0290f941e80a542b64 : FAILURE in 12m 14s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/42be5cf02e94421293fdea3d829462fd : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":4},{"id":"150aa006a53ecf68d780e33197e122cefe2e1fc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 01:41:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"48847afd270d2eb6e45542a03c3cfcdc001d1978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 02:43:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70c140813b24a60a342a0961c041bc1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ffe80c4e3e0404db0ed3031f7827b63 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be15af1164c549c9994936f23163d96c : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eba5129ee16c46609304fd49e8cf7808 : SUCCESS in 2m 54s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4e2d20ee272140479a591a5c6e944643 : SUCCESS in 52m 02s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1705d8b9b3e547d7b48d79ba40b3bd7e : SUCCESS in 36m 01s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/448284478b6f44f2b8bcc82657a3346d : SUCCESS in 56m 22s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/54c31ba0333541c5ac1c13fb00985fcb : SUCCESS in 44m 10s","accounts_in_message":[],"_revision_number":5},{"id":"23162438d5c63a006b0b7b52c504699381ed65e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-16 15:04:40.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":6},{"id":"dc664354747b00bd72290db0161a8c46d8273181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:17:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2725ac727c5b446fb0c07a3fcdaed637\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a39f33062a2044fd89f3aa5eafaad200 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3590782e1d504e8f8af2e393f0ad42cd : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b94b2558b3341a9a04b2232bd86ee32 : SUCCESS in 3m 58s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ff1f4e6676184ca1972a242e5b34b260 : SUCCESS in 39m 02s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/961f5a2fb29b46228380461f9d5e904c : SUCCESS in 53m 49s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f3229894ffdc4895b203274b6e56a731 : SUCCESS in 1h 01m 17s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d6fae8d3780c44f38b1cc05ef555913a : SUCCESS in 58m 46s","accounts_in_message":[],"_revision_number":6},{"id":"8def389b0d31af09dfa2b936ebd63213dc33c463","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-17 08:42:58.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2771f068b88206f3e08cd08d04054d91fd715f35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 08:43:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fbdf9c71c010f2aef3439fdfbf873bc9a648ab3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 10:01:41.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf8ec0684e87451fb7c3a4b9760eef63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aebbae874c5e4d529059940faf6e1b93 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a4d0cc2f0ed4a058f92855efd53d7a8 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f58b66a6df8417b9fdc415707cf59c2 : SUCCESS in 3m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5566f2f3eda44d638def61b6da44cdce : SUCCESS in 50m 54s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/904ad83b1897492f843cf9c6621a48b1 : SUCCESS in 32m 18s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e3c96c1754364369a8e48995120bddcd : SUCCESS in 40m 49s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/123da5ba459446219a93daa05ce69a00 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":6},{"id":"cf4c7beb157560d9f546c81bfd5ab190837b4c7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 10:01:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6d3baec5d0894de17df9207302b35b382b931f73","revisions":{"640301b676048b29dbdae92a2e12c7fdd6de4a7a":{"kind":"REWORK","_number":1,"created":"2024-11-18 06:56:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/935480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/935480/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/935480/1"}}},"commit":{"parents":[{"commit":"e82988b461a24c6d2436a27d86336db8901b1ebe","subject":"Select correct memcache server value according to IP version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e82988b461a24c6d2436a27d86336db8901b1ebe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:55:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:55:00.000000000","tz":540},"subject":"Enable caching","message":"Enable caching\n\nEnable caching in heat, keystone and nova. These services are known to\nleverage caching for effective performance improvement.\n\nChange-Id: Ibbcdb9a46bf02057a3640d8260f76a13f94e3882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/640301b676048b29dbdae92a2e12c7fdd6de4a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/640301b676048b29dbdae92a2e12c7fdd6de4a7a"}]},"branch":"refs/heads/master"},"9726f9434138ec37d85f0c879f616f913379edb2":{"kind":"REWORK","_number":2,"created":"2024-11-18 06:59:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/935480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/935480/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/935480/2"}}},"commit":{"parents":[{"commit":"e82988b461a24c6d2436a27d86336db8901b1ebe","subject":"Select correct memcache server value according to IP version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e82988b461a24c6d2436a27d86336db8901b1ebe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:55:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:58:23.000000000","tz":540},"subject":"Enable caching","message":"Enable caching\n\nEnable caching in heat, keystone and nova. These services are known to\nleverage caching for effective performance improvement.\n\nChange-Id: Ibbcdb9a46bf02057a3640d8260f76a13f94e3882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9726f9434138ec37d85f0c879f616f913379edb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9726f9434138ec37d85f0c879f616f913379edb2"}]},"branch":"refs/heads/master"},"04601e2519614b572b92949d6077eae8c81b1933":{"kind":"REWORK","_number":3,"created":"2024-11-18 14:25:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/935480/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/935480/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/935480/3"}}},"commit":{"parents":[{"commit":"e82988b461a24c6d2436a27d86336db8901b1ebe","subject":"Select correct memcache server value according to IP version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e82988b461a24c6d2436a27d86336db8901b1ebe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:55:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 14:15:14.000000000","tz":540},"subject":"Enable caching","message":"Enable caching\n\nEnable caching in heat, keystone and nova. These services are known to\nleverage caching for effective performance improvement.\n\nChange-Id: Ibbcdb9a46bf02057a3640d8260f76a13f94e3882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/04601e2519614b572b92949d6077eae8c81b1933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/04601e2519614b572b92949d6077eae8c81b1933"}]},"branch":"refs/heads/master"},"bba5c9b1a496f4977118df43682cb17b28f3737b":{"kind":"REWORK","_number":4,"created":"2024-11-18 15:02:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/935480/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/935480/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/935480/4"}}},"commit":{"parents":[{"commit":"e82988b461a24c6d2436a27d86336db8901b1ebe","subject":"Select correct memcache server value according to IP version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e82988b461a24c6d2436a27d86336db8901b1ebe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:55:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 15:00:47.000000000","tz":540},"subject":"Enable caching","message":"Enable caching\n\nEnable caching in heat, keystone and nova. These services are known to\nleverage caching for effective performance improvement.\n\nChange-Id: Ibbcdb9a46bf02057a3640d8260f76a13f94e3882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bba5c9b1a496f4977118df43682cb17b28f3737b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bba5c9b1a496f4977118df43682cb17b28f3737b"}]},"branch":"refs/heads/master"},"4fee8b2edf82fdcee48021bca78d6fab2a27ae8d":{"kind":"REWORK","_number":5,"created":"2024-11-19 01:41:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/935480/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/935480/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/935480/5"}}},"commit":{"parents":[{"commit":"e82988b461a24c6d2436a27d86336db8901b1ebe","subject":"Select correct memcache server value according to IP version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e82988b461a24c6d2436a27d86336db8901b1ebe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:55:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-19 01:40:50.000000000","tz":540},"subject":"Enable caching","message":"Enable caching\n\nEnable caching in heat, keystone and nova. These services are known to\nleverage caching for effective performance improvement.\n\nChange-Id: Ibbcdb9a46bf02057a3640d8260f76a13f94e3882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4fee8b2edf82fdcee48021bca78d6fab2a27ae8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4fee8b2edf82fdcee48021bca78d6fab2a27ae8d"}]},"branch":"refs/heads/master"},"6d3baec5d0894de17df9207302b35b382b931f73":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-12-16 15:04:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/80/935480/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/80/935480/6","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/80/935480/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/80/935480/6"}}},"commit":{"parents":[{"commit":"3385a0c0875a165e5320d8372e20e92dd02158de","subject":"Merge \"Select correct memcache server value according to IP version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3385a0c0875a165e5320d8372e20e92dd02158de"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:55:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-16 15:04:40.000000000","tz":0},"subject":"Enable caching","message":"Enable caching\n\nEnable caching in heat, keystone and nova. These services are known to\nleverage caching for effective performance improvement.\n\nChange-Id: Ibbcdb9a46bf02057a3640d8260f76a13f94e3882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6d3baec5d0894de17df9207302b35b382b931f73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6d3baec5d0894de17df9207302b35b382b931f73"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
