)]}'
{"id":"x%2Fpackstack~935481","triplet_id":"x%2Fpackstack~master~I4a302a731eed87e1202e8db014c908c9976492bd","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-10 18:46:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a302a731eed87e1202e8db014c908c9976492bd","subject":"nova: Drop unused metadata manifest","status":"MERGED","created":"2024-11-18 06:57:11.000000000","updated":"2024-12-10 18:46:39.000000000","submitted":"2024-12-10 18:46:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935481","meta_rev_id":"f86ea36a679a92d53ec566064f140e83cd40f600","_number":935481,"virtual_id_number":935481,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 18:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":2,"date":"2024-11-20 08:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":1,"date":"2024-12-10 17:29:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-18 08:02:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 08:29:11.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"1a47577b7d65bdc13cbe181fd901a152d90256a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 06:57:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa337626bb69ba093cdb71b37b509c0ff87b5bc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 06:57:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6286f7f71e527282c7ede1f34aa664704e3dd6bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 08:02:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d7882c053ec42cea0e7c1e019756da0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5165a5440b4cf082c6b6ffd6cbb265 : SUCCESS in 2m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a119cdc07874429947a5f6e88966b52 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f1554ad8af484ce3971f94c7737bd2ce : SUCCESS in 4m 21s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d98ebfd5e4d54e60900bb9f3407a9b38 : SUCCESS in 33m 09s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d8afa75fdafc4b139598b6606804bf54 : SUCCESS in 39m 27s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b6664afa15e5413e93c31f2cd63a4c8d : SUCCESS in 34m 21s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/55be3447bb7c46be922c04f1473d96ff : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":2},{"id":"3192730605b934f7e793438a8e00c8e8934a8a57","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-11-20 08:29:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"919bed9a2a2d372f0f463f15563f6e9674c096b1","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-10 17:29:01.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fcddf87a11220d735572b21568a9c8c866a951d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 17:29:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e418cc454ef5fcc63fbd62ab2b7fdcf1f1ec8557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 18:46:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdfda88224d74a1bb6e1950db8400204\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07308774522f451fba36d3d2202f687e : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9808d0b2703d49d1b8114387006e97be : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e4aaa4767a8454ca84257c2bc59abe2 : SUCCESS in 3m 17s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/3c9ec12e88684d74be9991f98df57d1e : SUCCESS in 1h 00m 08s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e8f36de282654df393951ed53df72a5a : SUCCESS in 33m 25s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/99fef64b6f734115a5b8b2e6bf24bcf0 : SUCCESS in 42m 56s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c0e14c0198a84f1e910c1878ebc19a8d : SUCCESS in 46m 47s","accounts_in_message":[],"_revision_number":2},{"id":"f86ea36a679a92d53ec566064f140e83cd40f600","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 18:46:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9b73a93ae3d0b4620fde296ab732089aedc7e487","revisions":{"4e9f9d43194911e9cda3d3e00cc042678fc733f2":{"kind":"REWORK","_number":1,"created":"2024-11-18 06:57:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/935481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/81/935481/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/81/935481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/81/935481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/81/935481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/81/935481/1"}}},"commit":{"parents":[{"commit":"640301b676048b29dbdae92a2e12c7fdd6de4a7a","subject":"Enable caching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/640301b676048b29dbdae92a2e12c7fdd6de4a7a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:56:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:56:52.000000000","tz":540},"subject":"nova: Drop unused metadata manifest","message":"nova: Drop unused metadata manifest\n\nChange-Id: I4a302a731eed87e1202e8db014c908c9976492bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4e9f9d43194911e9cda3d3e00cc042678fc733f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4e9f9d43194911e9cda3d3e00cc042678fc733f2"}]},"branch":"refs/heads/master"},"9b73a93ae3d0b4620fde296ab732089aedc7e487":{"kind":"REWORK","_number":2,"created":"2024-11-18 06:57:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/935481/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/81/935481/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/81/935481/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/81/935481/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/81/935481/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/81/935481/2"}}},"commit":{"parents":[{"commit":"120c142c1dea4cca18fe7518a0c12f653067ae93","subject":"Merge \"Create volume types by built-in logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/120c142c1dea4cca18fe7518a0c12f653067ae93"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:56:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:57:37.000000000","tz":540},"subject":"nova: Drop unused metadata manifest","message":"nova: Drop unused metadata manifest\n\nChange-Id: I4a302a731eed87e1202e8db014c908c9976492bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9b73a93ae3d0b4620fde296ab732089aedc7e487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9b73a93ae3d0b4620fde296ab732089aedc7e487"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
