)]}'
{"id":"x%2Fpackstack~937784","triplet_id":"x%2Fpackstack~stable%2F2023.2~I12012eaaee9b26db03105e38545340dc317cd6d4","project":"x/packstack","branch":"stable/2023.2","topic":"remove-definition-of-retired-projects","attention_set":{"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-12-16 15:44:15.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I12012eaaee9b26db03105e38545340dc317cd6d4","subject":"Remove definition of retired puppet projects","status":"NEW","created":"2024-12-16 15:17:20.000000000","updated":"2024-12-17 09:30:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b6b7920d2f5b7acc21b167c9b85f3372cb42b6a9","_number":937784,"virtual_id_number":937784,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-17 09:30:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-16 15:44:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e1145eebc8521b237b183fc973f1232a3221ea8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-16 15:17:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4845db10d3e87167e5f26fd5f37310ccf50f7037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 15:44:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b24c0ae643ba472a92f70df7bde93c83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eabeec818d6c4d09ac43459836db852d : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ee97e390d6d460cb264518e71ab05cf : SUCCESS in 2m 57s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ad3d2ad0147343138111086835e9a335 : FAILURE in 13m 59s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bbd8b98f6caa4c468ddf43ead9e69276 : FAILURE in 10m 46s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/10b08c63259c42919b917d4d0477b766 : FAILURE in 9m 56s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a12c956d29744162b7f6a0e2d7411b14 : FAILURE in 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1b482c6109337ada650c19e60e332564f4c2b953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-17 08:50:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b6b7920d2f5b7acc21b167c9b85f3372cb42b6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 09:30:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c1eab426c6d495fac684158f09bb26c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/130d60de848841759f91b5b7264746be : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/653627299b6c4321aee9654a254dfb47 : SUCCESS in 3m 19s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6227db5181d64a15bee5228c12f6b1f0 : FAILURE in 22m 43s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/a749defff7ec4265acab1301754d6e03 : FAILURE in 30m 01s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2831fef913ee4735bf18dcdbb5e3a5bd : FAILURE in 27m 53s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1e15207985ec4729a30b04460325cf5a : FAILURE in 17m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c25bc01cad1fb0e740d471f2a20f3342eefb5cbc","revisions":{"9cdfd85d3ca5d0bb9d62b51a25af87e139721035":{"kind":"REWORK","_number":1,"created":"2024-12-16 15:17:20.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/84/937784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/84/937784/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/84/937784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/84/937784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/84/937784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/84/937784/1"}}},"commit":{"parents":[{"commit":"cd5411000558c78eb0075f210bffa081afba567f","subject":"Switch to stable/2023.2 branh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cd5411000558c78eb0075f210bffa081afba567f"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 15:17:01.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 15:17:01.000000000","tz":60},"subject":"Remove definition of retired puppet projects","message":"Remove definition of retired puppet projects\n\nChange-Id: I12012eaaee9b26db03105e38545340dc317cd6d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9cdfd85d3ca5d0bb9d62b51a25af87e139721035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9cdfd85d3ca5d0bb9d62b51a25af87e139721035"}]},"branch":"refs/heads/stable/2023.2"},"c25bc01cad1fb0e740d471f2a20f3342eefb5cbc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-17 08:50:56.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/84/937784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/84/937784/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/84/937784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/84/937784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/84/937784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/84/937784/2"}}},"commit":{"parents":[{"commit":"5996aa3f609f7a1f06a7be6a5444649f40160e36","subject":"Fix branch to pull puppetlabs-xinetd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5996aa3f609f7a1f06a7be6a5444649f40160e36"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 15:17:01.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-17 08:50:56.000000000","tz":0},"subject":"Remove definition of retired puppet projects","message":"Remove definition of retired puppet projects\n\nChange-Id: I12012eaaee9b26db03105e38545340dc317cd6d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c25bc01cad1fb0e740d471f2a20f3342eefb5cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c25bc01cad1fb0e740d471f2a20f3342eefb5cbc"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
