)]}'
{"id":"x%2Fpackstack~937785","triplet_id":"x%2Fpackstack~stable%2F2023.1~I95310c6b961eab2f251b1ddf3218f202a79a3de8","project":"x/packstack","branch":"stable/2023.1","topic":"remove-definition-of-retired-projects","attention_set":{},"removed_from_attention_set":{"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-12-17 08:42:23.000000000","reason":"\u003cGERRIT_ACCOUNT_31068\u003e replied on the change","reason_account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}},"hashtags":[],"change_id":"I95310c6b961eab2f251b1ddf3218f202a79a3de8","subject":"Remove definition of retired puppet projects","status":"MERGED","created":"2024-12-16 15:17:57.000000000","updated":"2024-12-17 09:50:21.000000000","submitted":"2024-12-17 09:50:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937785-remove-definition-of-retired-projects","meta_rev_id":"ec58a79a508011d45db4bda41851ed67388f7147","_number":937785,"virtual_id_number":937785,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-17 09:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2024-12-17 07:28:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-12-17 08:42:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-12-17 08:42:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-16 17:01:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-17 07:28:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"bcc38bfc135adb62f77014f2c661e66096773333","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-16 15:17:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3233ffd6df221e4d177a5b6c18f6e51662236d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 17:01:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/321dc47d6b714a819f988bc91ef71384\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5bdc933a75452abbafb635a3fd8ac5 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be308c2805a749ec8b9cc3d13f4f6dd6 : SUCCESS in 3m 27s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/db289d472955489cbbd608a0c02479b9 : SUCCESS in 38m 06s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/eafb2d33f52245e5a312e4172d8df10c : SUCCESS in 38m 51s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6198bfd3907242c299820a0d3895ff8a : SUCCESS in 41m 14s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/eee8acd5064d48939a63e361e016a350 : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":1},{"id":"636c8d119c15fb5f6301eda36a6f1bb4a56f7138","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-12-17 07:28:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"81dd75b62197dca9fd0ca056ba7615290aeb14ab","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-17 08:42:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ee6785082bfd38b5f505247eb610ee7db91c9dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 08:43:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"338fb427367d80bdb77a9282aba6e4802f3c11e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 09:50:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/965f5c4e5fe44c27956459b5645aa32b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5f431992734f4f86e96629d1d218db : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efb90e7bd6184af9be45c2f993bcec26 : SUCCESS in 3m 03s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9f69ac844eb84525a3e7935871f1d748 : SUCCESS in 55m 06s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/a1207f70ee3848aba7cba60429fea1b7 : SUCCESS in 50m 17s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/595064b45ee844aba0bf0a9ce4a85ca1 : SUCCESS in 1h 00m 31s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bf85ccd345f44e719d674ebed395f0c1 : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ec58a79a508011d45db4bda41851ed67388f7147","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 09:50:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0194ffd2bccd757f6c5ace6e14de58c3e2222fa","revisions":{"b0194ffd2bccd757f6c5ace6e14de58c3e2222fa":{"kind":"REWORK","_number":1,"created":"2024-12-16 15:17:57.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/85/937785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/85/937785/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/85/937785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/85/937785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/85/937785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/85/937785/1"}}},"commit":{"parents":[{"commit":"a0d77af51ca2440c419311faf0f1fd0cc7df4ca5","subject":"Add enable-chassis-as-gw option for OVN CMS cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a0d77af51ca2440c419311faf0f1fd0cc7df4ca5"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 15:17:52.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-12-16 15:17:52.000000000","tz":60},"subject":"Remove definition of retired puppet projects","message":"Remove definition of retired puppet projects\n\nChange-Id: I95310c6b961eab2f251b1ddf3218f202a79a3de8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b0194ffd2bccd757f6c5ace6e14de58c3e2222fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b0194ffd2bccd757f6c5ace6e14de58c3e2222fa"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
