)]}'
{"id":"x%2Fpackstack~937853","triplet_id":"x%2Fpackstack~master~I462da9e4e387cc43b6dfd0ae7b677e06b38e0e3d","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-24 10:58:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I462da9e4e387cc43b6dfd0ae7b677e06b38e0e3d","subject":"Remove error log pattern for swift3 installation","status":"MERGED","created":"2024-12-17 09:41:11.000000000","updated":"2024-12-24 10:58:07.000000000","submitted":"2024-12-24 10:58:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937853","meta_rev_id":"a6350db666d40d5ab28a85f68de873421810d7b0","_number":937853,"virtual_id_number":937853,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-24 10:58:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":2,"date":"2024-12-24 09:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":1,"date":"2024-12-24 09:55:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-17 10:47:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-24 09:55:39.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"a781046cc3d779dddce0ccebb766fc275cf716e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-17 09:41:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe2e841b05d37605857e7188ccf662abcadd34a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 10:47:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06aa1f05710e47a8a24cb3d655c3d34e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef264269d74143ec8da6e95ba8d43ad6 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/023f0cde63664b74a1599d2b16d21d2b : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/405ee09d02764b7ea30af6be54f59f83 : SUCCESS in 3m 24s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6cc232be34954403a5cbcf8f6dcc32f6 : SUCCESS in 41m 52s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5931dcbe63cd49cbbbe492c9f144468a : SUCCESS in 35m 13s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/922a97c739374b5582de469b55b0dde3 : SUCCESS in 58m 20s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f946c3350d5d41f1aefb35eb237d4f70 : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":1},{"id":"10bc30091c753d74114d0e5222dc7042ba19fe50","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-24 09:55:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca631c9e06db38704c47051a85986b329e31cad7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 09:56:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"997bbf535a119595595f92d0e9688ba32b6b6dfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 10:58:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d8d1d3f5ebf47c8a870983d0f902243\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a84e1e080df3468b9bff65eb2190b32c : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cb0a357b7fe48629cb5861f315dd21c : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aac2da9bb08f43d68365bf402cb39061 : SUCCESS in 3m 16s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e8ea83e497ac4909a90cad5b9886cc0d : SUCCESS in 36m 38s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0930d40e0c8745a0aefc70786414245f : SUCCESS in 29m 11s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9e078cfdd3ca4409b944f299ed2d926c : SUCCESS in 30m 36s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7531fdf6a3104af88bbb9766b1842ca5 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a6350db666d40d5ab28a85f68de873421810d7b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 10:58:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4313448e986c0b61fe6764db135ec6c6b99be399","revisions":{"4313448e986c0b61fe6764db135ec6c6b99be399":{"kind":"REWORK","_number":1,"created":"2024-12-17 09:41:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/937853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/53/937853/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/53/937853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/53/937853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/53/937853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/53/937853/1"}}},"commit":{"parents":[{"commit":"fb7aada21c65c896db911be08dc7386ef5442e8d","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fb7aada21c65c896db911be08dc7386ef5442e8d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-17 09:40:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-17 09:40:05.000000000","tz":540},"subject":"Remove error log pattern for swift3 installation","message":"Remove error log pattern for swift3 installation\n\nThe package name for RHEL was fixed by [1]. Also swift3 support was\nremoved by [2].\n\n[1] e6ba75a1618c580e48cb9830898194cdbf38c13e\n[2] d4ec1b631faa00aed2a939dcf2c76217024537ff\n\nChange-Id: I462da9e4e387cc43b6dfd0ae7b677e06b38e0e3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4313448e986c0b61fe6764db135ec6c6b99be399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4313448e986c0b61fe6764db135ec6c6b99be399"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
