)]}'
{"id":"x%2Fpackstack~938475","triplet_id":"x%2Fpackstack~master~I0b4a08516d93cbb8bb62551e629ccdcf04ca62c1","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-08 13:12:20.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2025-01-08 12:07:34.000000000","reason":"\u003cGERRIT_ACCOUNT_31068\u003e replied on the change","reason_account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}},"hashtags":[],"change_id":"I0b4a08516d93cbb8bb62551e629ccdcf04ca62c1","subject":"Remove usage of the keystone client class","status":"MERGED","created":"2025-01-06 13:24:22.000000000","updated":"2025-01-08 13:12:20.000000000","submitted":"2025-01-08 13:12:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938475","meta_rev_id":"e6de8c271ca5d295c4c34b108a8a43ff2b015172","_number":938475,"virtual_id_number":938475,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-08 13:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":2,"date":"2025-01-08 12:07:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-08 10:19:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":1,"date":"2025-01-08 12:07:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-06 14:05:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-08 10:19:54.000000000","updated_by":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"reviewer":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"state":"REVIEWER"},{"updated":"2025-01-08 10:21:29.000000000","updated_by":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"2ef17c6d261953d2c22261c212f898f96e3dab96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2025-01-06 13:24:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf2ea810922738addb014b0281fa62068e84877c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 14:05:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9db47175f4644aeabad57e6ff8a6de5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a3b4cd9e9540069a8903ad42ac479b : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fe15b36499a466cba08f31003313d3b : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a84475cb423545039b3961c3a62e7885 : SUCCESS in 3m 43s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5e53bfeb644e46ce89e1a099fa166f01 : SUCCESS in 37m 13s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/78d079bc1f41451ca1d076587ab56366 : SUCCESS in 27m 10s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/819f29c01e9f48b8bd9456be14cc8979 : SUCCESS in 39m 34s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bbb4ea51721b421cbfefb4b0c7f1786e : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":1},{"id":"fbaf1589df714b0a872c23c17f33041475b46f9f","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2025-01-08 10:19:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b396fda7162892e8511065b8f0ba1fa7c761c77a","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2025-01-08 12:07:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"226dffc4206e7d2df938d601ba7aa8689df6eb59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 12:08:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bea55cfef938b7ed31f2e6b517441f83de90770","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 13:12:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c23410e2c1a4e2f9704e9d8b2fa0de9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/765c64f1acec48ffabcc7c6ce431d59b : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6079395425e4cdc862b64a6cebabb51 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c185b537c19541e786fa3cd3f1cdf021 : SUCCESS in 3m 27s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/da373073ccc54bb99473ace1c8e34123 : SUCCESS in 49m 43s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/62339299382c46b0bd9cdbb7c5902cb5 : SUCCESS in 49m 46s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1144b4982e384b04915c1410b1801239 : SUCCESS in 55m 18s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/dfe9fce31c68437fad0bc097aaf733b5 : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e6de8c271ca5d295c4c34b108a8a43ff2b015172","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 13:12:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8588829b4a60bc6b982bf83549eacfdad10e8f1b","revisions":{"8588829b4a60bc6b982bf83549eacfdad10e8f1b":{"kind":"REWORK","_number":1,"created":"2025-01-06 13:24:22.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/75/938475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/75/938475/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/75/938475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/75/938475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/75/938475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/75/938475/1"}}},"commit":{"parents":[{"commit":"336a25b3e089bde477e8144f0828c0da5b964fd6","subject":"Merge \"Remove error log pattern for swift3 installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/336a25b3e089bde477e8144f0828c0da5b964fd6"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2025-01-06 13:22:50.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2025-01-06 13:22:54.000000000","tz":60},"subject":"Remove usage of the keystone client class","message":"Remove usage of the keystone client class\n\nThe class was deprecated long time ago and removed\nrecently [1]\n\n[1] f271472b4824cdad1d0939c7c2d3a68556d1467d\n\nChange-Id: I0b4a08516d93cbb8bb62551e629ccdcf04ca62c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8588829b4a60bc6b982bf83549eacfdad10e8f1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8588829b4a60bc6b982bf83549eacfdad10e8f1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
