)]}'
{"id":"x%2Fpackstack~938859","triplet_id":"x%2Fpackstack~master~Ibce1d54ef9c26329eca23d70e8265d32e6d78caf","project":"x/packstack","branch":"master","topic":"packstack-constraints","attention_set":{},"removed_from_attention_set":{"37373":{"account":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"last_update":"2025-01-10 08:21:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibce1d54ef9c26329eca23d70e8265d32e6d78caf","subject":"Updated from Packstack modules constraints","status":"MERGED","created":"2025-01-10 00:02:44.000000000","updated":"2025-01-10 08:21:54.000000000","submitted":"2025-01-10 08:21:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938859-packstack-constraints","meta_rev_id":"72f16bb4ef3217440d9e3cbfa3f85a6a7a6fdc15","_number":938859,"virtual_id_number":938859,"owner":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 08:21:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":2,"date":"2025-01-10 07:21:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":1,"date":"2025-01-10 07:21:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 01:10:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 07:21:54.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"2ec1ce490a42afb6309b356cf94f9d78db8e6d2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2025-01-10 00:02:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c932a77095ffebb7ab1b9414b4431b4e0cf8b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 01:10:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65d05869c5b145d5915145172c669d47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5614265fa1945e2a1e6b34e0230cd28 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08331fd4f4d049ca86c12386c4093c87 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a40e4b4cb18498bb02dd65c48baa4cb : SUCCESS in 3m 04s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e032411ba5424323819e99c80f3d3fe3 : SUCCESS in 33m 56s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/35cb5057637a472d8391051c7244d35a : SUCCESS in 33m 48s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8b9cd36eecfe4897a9679e8ff14b8fb6 : SUCCESS in 37m 53s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5d2f55f1479a4c5c9446454ac8bd6dad : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5a356729c7e68cecff2f993c3be4850e4a7f6dd8","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2025-01-10 07:21:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81b1ce611053d19decf4e3c89e5398d4bf353142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 07:22:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9fe324828bf172be6c6a27306e4d486cd94d9103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 08:21:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31f267fa68874bcba82cf71737d89c51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a1173c4f06c4d60acc652c155be6381 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2392561fbe914150b5893631c9252a5a : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a35dd2346b4648b9972050ea7c81ff91 : SUCCESS in 3m 32s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5fb74b7a6d0a4896ab43d41122c447e9 : SUCCESS in 33m 34s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b644183e196941a2bc8f71f176c577c9 : SUCCESS in 31m 13s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/fddc57516ec641b7aef9c6964b430411 : SUCCESS in 42m 02s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f2f869391b8f42e0bcfe0bfd34422e49 : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":1},{"id":"72f16bb4ef3217440d9e3cbfa3f85a6a7a6fdc15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 08:21:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9d49fb9b61f8f27ef76e478e7020ed855ec124b","revisions":{"c9d49fb9b61f8f27ef76e478e7020ed855ec124b":{"kind":"REWORK","_number":1,"created":"2025-01-10 00:02:44.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/59/938859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/59/938859/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/59/938859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/59/938859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/59/938859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/59/938859/1"}}},"commit":{"parents":[{"commit":"8588829b4a60bc6b982bf83549eacfdad10e8f1b","subject":"Remove usage of the keystone client class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8588829b4a60bc6b982bf83549eacfdad10e8f1b"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-01-10 00:02:44.000000000","tz":-300},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-01-10 00:02:44.000000000","tz":-300},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: Ibce1d54ef9c26329eca23d70e8265d32e6d78caf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c9d49fb9b61f8f27ef76e478e7020ed855ec124b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c9d49fb9b61f8f27ef76e478e7020ed855ec124b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
