)]}'
{"id":"x%2Fpackstack~939188","triplet_id":"x%2Fpackstack~master~Iadcf0ffe4b796ac60c1a23dce7dba2797757fc92","project":"x/packstack","branch":"master","topic":"packstack-constraints","attention_set":{},"removed_from_attention_set":{"37373":{"account":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"last_update":"2025-01-15 09:02:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iadcf0ffe4b796ac60c1a23dce7dba2797757fc92","subject":"Updated from Packstack modules constraints","status":"MERGED","created":"2025-01-14 00:04:58.000000000","updated":"2025-01-15 09:02:57.000000000","submitted":"2025-01-15 09:02:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939188-packstack-constraints","meta_rev_id":"2fd37df1b5d59462e674a89c7396a1ef23e0aed6","_number":939188,"virtual_id_number":939188,"owner":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-15 09:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":2,"date":"2025-01-15 07:57:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":1,"date":"2025-01-15 07:57:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 01:10:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-14 08:03:32.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"CC"},{"updated":"2025-01-15 07:57:34.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"7052a1b2c621cc29e94c33dd00977883b349747f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2025-01-14 00:04:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b1f8deecac7730ed40ff5a1b46eaeae6cc20298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 01:10:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c3c203761314b3a9b695801032e068a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb47790deef44cd4997b7bda07e550b4 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50e95262ce6841e7ad8d2bf0b0db438f : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf29c87011bc4f818bb0d03ea3d91210 : SUCCESS in 3m 01s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fb2838b80db34e13b21150cdebb87758 : SUCCESS in 54m 54s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9aa025aacd19419493cf6e50783600df : FAILURE in 9m 51s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9ec631c48cd449d3ad52dff2c0613e17 : SUCCESS in 59m 44s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f854098107ce4b3c814b1ba5634d3afd : FAILURE in 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"73a047f3d431ca29976c393647db87eaab5a2139","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2025-01-14 08:03:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"856b444012f7c913c8b9910eb77fd99776343f53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 09:08:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b2cb9e3b3d749958a15f77eec833d2d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/878ec77aef884b3d9a77bc33c3f73c0e : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/583a227ebbdc4542bd210431a67d7dd6 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/478b966cd03942a1abc1ed78217fd898 : SUCCESS in 2m 54s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4bd181a76a8448b99da070dd39d0e937 : SUCCESS in 27m 11s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d34abfe429cf4d2291ea93da54a42f8c : SUCCESS in 47m 56s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/80fc77d9cd6c4f249b22f8c37d5c1e10 : SUCCESS in 41m 36s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/433e1219f81646899725916bffe866c1 : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bd62a6c928b11afe42bb35ceeda2003229807041","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2025-01-15 07:57:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fd90f16037323736ad66673e328fe45e0eab26e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 07:58:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ce570a9050e19d72ced84c1e50796329092ffe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 09:02:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5762d166318849dca215f4fd25544350\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47de1c111db9498eb2bcec1d1501f9c2 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c492ac32e22b423cb017a169bc025d66 : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c62361a6e8ca42ac86a56e054e7684d9 : SUCCESS in 2m 58s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e78cac876a5742c689ba713dbb55d5a6 : SUCCESS in 52m 35s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5e5a6d716f3e415d92704877e1225cb4 : SUCCESS in 26m 55s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/63878b0da97744af998012f98e32af9e : SUCCESS in 38m 40s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b7bb1c82c2b84692b693b4e73ee94569 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2fd37df1b5d59462e674a89c7396a1ef23e0aed6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 09:02:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9073cffbd530b62aab18ee2350bd0bdd7fec565e","revisions":{"9073cffbd530b62aab18ee2350bd0bdd7fec565e":{"kind":"REWORK","_number":1,"created":"2025-01-14 00:04:58.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/88/939188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/88/939188/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/88/939188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/88/939188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/88/939188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/88/939188/1"}}},"commit":{"parents":[{"commit":"c9d49fb9b61f8f27ef76e478e7020ed855ec124b","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c9d49fb9b61f8f27ef76e478e7020ed855ec124b"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-01-14 00:04:58.000000000","tz":-300},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-01-14 00:04:58.000000000","tz":-300},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: Iadcf0ffe4b796ac60c1a23dce7dba2797757fc92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9073cffbd530b62aab18ee2350bd0bdd7fec565e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9073cffbd530b62aab18ee2350bd0bdd7fec565e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
