)]}'
{"id":"x%2Fpackstack~943379","triplet_id":"x%2Fpackstack~master~I1ee2084d9308d8c720a83411993267a82959a6bf","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-06 00:59:08.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2025-03-06 01:54:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1ee2084d9308d8c720a83411993267a82959a6bf","subject":"Fix pretty outdated manila service image location","status":"MERGED","created":"2025-03-05 12:00:38.000000000","updated":"2025-03-06 01:54:29.000000000","submitted":"2025-03-06 01:54:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943379","meta_rev_id":"fb3d2c866e0b9a72496c89ed9b423316c205d8f9","_number":943379,"virtual_id_number":943379,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 01:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2025-03-05 16:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-05 13:23:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2025-03-06 00:59:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 13:03:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-05 13:23:45.000000000","updated_by":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"reviewer":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"state":"REVIEWER"},{"updated":"2025-03-05 14:37:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2025-03-05 16:00:31.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"3abb6e60f08e4aa279aecb5b7c357f874dabf9e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-05 12:00:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8db01192c620e98b701fe681cf155df7ee795f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 13:03:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ea3bd20892641b2807aaf36c167919d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94fb1debf5a14557b01ed2a7e4d7f38c : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdd1c2be300b4ed2910986ac575ef5c7 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1db578489e7f42f7a04bc910f56ad00c : SUCCESS in 4m 03s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6727edae09bd4e1eb3272e72becb9e73 : SUCCESS in 55m 45s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f695f2406d434417938b75250de77b81 : SUCCESS in 26m 20s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a1affb29f054430586254505efab4664 : SUCCESS in 37m 15s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e31484002f564b059515568e7ab9c197 : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e94d0fab04963299d4047673b27189c37a41faf1","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2025-03-05 13:23:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"186620c8e3c8bda7de5cd007ec52f281c6ddd99b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-03-05 16:00:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a37e177c0725b856cf055bbf4b8c1542fbb6004","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-06 00:59:08.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba6bbbea72fda2bb0670cb72ea7f134d339fe788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 00:59:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcfb6adf523a07a62a631f7ec19fcbb8b6ea3e87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 01:54:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aad4b990abb949d2a1289b1635597c39\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ac4791e35f4b2585ba3659675303a0 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/367806bec9424dcaae504758691553d9 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b47fb695b4b84148b2175d90f054cca1 : SUCCESS in 3m 52s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2071d318de1b448fa2ddde2e7197cd85 : SUCCESS in 40m 38s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3b51f87c3b654df89754d74f5ab5dbe3 : SUCCESS in 35m 32s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/caba73cc1bec4d5fb544d600aff2fbad : SUCCESS in 38m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/744e7da4041744558967da89b9d1151f : SUCCESS in 34m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fb3d2c866e0b9a72496c89ed9b423316c205d8f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 01:54:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46c0f5d3db96d8bc95b35cb7fb8a81b6ab38ac82","revisions":{"46c0f5d3db96d8bc95b35cb7fb8a81b6ab38ac82":{"kind":"REWORK","_number":1,"created":"2025-03-05 12:00:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/943379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/79/943379/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/79/943379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/79/943379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/79/943379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/79/943379/1"}}},"commit":{"parents":[{"commit":"11b668cc14db266d2694667c356ce6dc8c9d3277","subject":"swift: Remove redundant parameters for bulk/ratelimit middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/11b668cc14db266d2694667c356ce6dc8c9d3277"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-05 12:00:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-05 12:00:10.000000000","tz":540},"subject":"Fix pretty outdated manila service image location","message":"Fix pretty outdated manila service image location\n\nThe existing link no longer provides a valid content.\n\nChange-Id: I1ee2084d9308d8c720a83411993267a82959a6bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/46c0f5d3db96d8bc95b35cb7fb8a81b6ab38ac82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/46c0f5d3db96d8bc95b35cb7fb8a81b6ab38ac82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
