)]}'
{"id":"x%2Fpackstack~946220","triplet_id":"x%2Fpackstack~master~I181969ef8251a4c08e111ab824b26e4da1214ed8","project":"x/packstack","branch":"master","topic":"packstack-constraints","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-06-13 03:49:36.000000000","reason":"Change was submitted"},"37373":{"account":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"last_update":"2025-06-13 03:49:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I181969ef8251a4c08e111ab824b26e4da1214ed8","subject":"Updated from Packstack modules constraints","status":"MERGED","created":"2025-04-03 00:02:42.000000000","updated":"2025-06-13 03:49:36.000000000","submitted":"2025-06-13 03:49:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946220-packstack-constraints","meta_rev_id":"7bf8f288a681350815e3140fefe301a1976c9454","_number":946220,"virtual_id_number":946220,"owner":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-13 03:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-06-13 02:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-06-13 02:10:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-03 00:38:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-17 13:13:28.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-05-09 11:37:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-06-13 02:09:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"f2799c202048525f80c247d51a6469fa56c3974f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2025-04-03 00:02:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c45a7767d042fc7d62d0d23e93b91881da94eb92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 00:38:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48c33f6bc529452b8086b964f311030e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36509642662e4f86b9b38ba53c9a8bc0 : SUCCESS in 1m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df019df546374c8ea12fd99e8b8de8be : SUCCESS in 1m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25413132092c4165a9d612065c697b75 : SUCCESS in 1m 45s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b0ab727ff1c747eb8998cd92744fda16 : SUCCESS in 27m 48s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/19da3557f1df42a29560819fb6fb3692 : SUCCESS in 20m 19s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/19a27eef1f4f41d2bc8ec3a40578ae63 : SUCCESS in 34m 35s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1bc4107d9d6744efb2ca90deffa04666 : SUCCESS in 34m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6b8902d1633041e3cf79c990f270b2b44c3b81c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2025-04-17 00:02:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f25a0a8c064fd2eeeddd93ba268c1a32942ee20e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 00:48:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6590b4f9c9da4dfb924a8bc8c4c72747\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14b11510c8140088074063c12eea779 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a9be522f67e456687d3c50c3be89074 : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2012e5b1f140415292ef817afab08b07 : SUCCESS in 6m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a7c47a0b385c4b0ea7c5b29fa41b0f75 : SUCCESS in 37m 31s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/645fd9d3e4834e2db3c76a16d67c2e36 : SUCCESS in 23m 02s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5474f1cdf1c74af0bf1cf9762fd704fc : SUCCESS in 26m 45s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3d34102f6b4b4f2b9871a6af32212cad : FAILURE in 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"929560f3faf700356d3e4e003f14deab692dba88","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-17 13:13:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"590c1e1e635282dcb664e86a172a0abdba134c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 14:06:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c42bee82379479b8b14d800846376ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71931c757ff44c3f81fb847f45632fda : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/815371a5539e4854ae046af1419dac80 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eca2f4e62ce641efb7dbb4f4d8724fe0 : SUCCESS in 3m 49s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9dac17b34bdc48669c71e22a16fa2dac : SUCCESS in 36m 15s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/65c0261a65c94f2ca8744515b6441e85 : SUCCESS in 34m 43s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9057a93c2ac741e89bc9a33597057327 : SUCCESS in 47m 29s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/448a633656e5416d8a869acdd0408803 : SUCCESS in 44m 17s","accounts_in_message":[],"_revision_number":2},{"id":"1545f4f5a65bd01dd4d297545c10e662c2df0028","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2025-04-22 00:02:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a1d230f7243bb54bce40b41bfcf2ec8ecdf811f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 00:28:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c8758f07f64409bb72824086e56ae00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/227e798301ca4a25a4bf565dd164af03 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/873e0f18809a41f8a887673be63649cc : SUCCESS in 2m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9192f3adb23242da984021938827cc9d : SUCCESS in 1m 45s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bd90a40f8a8d48b38c4b90c4901fcb48 : SUCCESS in 20m 22s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6475b643376342aaa00401c48e36864b : SUCCESS in 24m 53s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/70380cfe13534c4abd55b3a9c81d972d : SUCCESS in 20m 36s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f54ee571a1a949159449673a2471a05b : FAILURE in 17m 17s","accounts_in_message":[],"_revision_number":3},{"id":"98cb52c1ce6d9cdfe3874b686149f9d4f59f82a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2025-05-09 00:02:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7a4179a3a71832dbe02ad579f6dc989de826724c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 00:51:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5e340b3400d46edaaa70829a7269176\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c328ef2a7f3e45ee8efb1d5aa445e8d6 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d75e0f54cd124bcfb701805568b526fb : SUCCESS in 2m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7444f24ce7343568d29b7bdb60c1622 : SUCCESS in 2m 44s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6d1238f45b364c7799bf4390097f331d : SUCCESS in 48m 48s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/44219082cf7e404ea00b9d125eada3a6 : SUCCESS in 24m 19s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0e727ebb74d94972b1d2c7381fc915d9 : SUCCESS in 36m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4890d3c08ab243a7b720f5804b9389b0 : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":4},{"id":"a8811dfedbc4166cdbaea36f4172825079ad22c4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-09 11:37:21.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4df0773fa3cf180618e3137e9540dc4e4c1a4e0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2025-06-13 00:02:29.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"fd3967937fcd4238e678eeb7830a5ae8f1bd7646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 00:57:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb5830b307644d2890e6051b3167f66d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9457629c71334299a3165b4ca15b7205 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e2b5a079e6b4b4fb86b40c7b1b6ba5b : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/92fd79edb8634ee3a9bb8ec70f282b25 : SUCCESS in 3m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/646a195204e24186a07d1747b0795065 : SUCCESS in 30m 57s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8e14cdd434b74c0dbb289b5df6355929 : SUCCESS in 34m 45s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d5550ace3d2948f0b0c8a3906032158c : SUCCESS in 38m 13s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/efa9f6cc7c0f4ceab54ec206b0fa8cd4 : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":5},{"id":"88b1f416e31b8b46cfc9157e6aeb28a1a2e25033","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-13 02:09:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"662c49b014ebfb03b379beb53560630b833a2126","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-13 02:10:09.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"fdedc5882a8cc9b78600f36105f2775e27271b63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 02:11:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e34c8800bdafa0e63b85d8d52a65e5bb54587e43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 03:49:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c95019ce262c4262810c0ea4c22722ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745babe6ac3140bb93364507aca33e41 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f75ab8c4b54848229251847368a9b5d1 : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfcc4faef53f4cef87880a35d707c312 : SUCCESS in 2m 53s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e1cd1fdee70c4244a96dd56bc8d78e83 : SUCCESS in 46m 03s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cd1c7063ac2f4bcd8f4dda44aa2ffd38 : SUCCESS in 19m 46s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a5462a0650984146874f2dc7520fcb57 : SUCCESS in 55m 59s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f53b46df05d14121a0b1baf063247198 : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":5},{"id":"7bf8f288a681350815e3140fefe301a1976c9454","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 03:49:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9d3df637902ec402e405c7e8586b5b0debabc987","revisions":{"f5622238f9a7a4d706d48c0dad6339bc34c83b95":{"kind":"REWORK","_number":1,"created":"2025-04-03 00:02:42.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/20/946220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/20/946220/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/20/946220/1"}}},"commit":{"parents":[{"commit":"d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb","subject":"Merge \"Parse json strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-04-03 00:02:41.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-04-03 00:02:41.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I181969ef8251a4c08e111ab824b26e4da1214ed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f5622238f9a7a4d706d48c0dad6339bc34c83b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f5622238f9a7a4d706d48c0dad6339bc34c83b95"}]},"branch":"refs/heads/master"},"f3b22e1f50e71ac58e52fc78e4957ceb969e51d2":{"kind":"REWORK","_number":2,"created":"2025-04-17 00:02:49.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/20/946220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/20/946220/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/20/946220/2"}}},"commit":{"parents":[{"commit":"d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb","subject":"Merge \"Parse json strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-04-03 00:02:41.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-04-17 00:02:49.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I181969ef8251a4c08e111ab824b26e4da1214ed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f3b22e1f50e71ac58e52fc78e4957ceb969e51d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f3b22e1f50e71ac58e52fc78e4957ceb969e51d2"}]},"branch":"refs/heads/master"},"d47266c82ab0ca238b908179103d0b8be2496f78":{"kind":"REWORK","_number":3,"created":"2025-04-22 00:02:17.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/20/946220/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/20/946220/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/20/946220/3"}}},"commit":{"parents":[{"commit":"d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb","subject":"Merge \"Parse json strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-04-03 00:02:41.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-04-22 00:02:16.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I181969ef8251a4c08e111ab824b26e4da1214ed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d47266c82ab0ca238b908179103d0b8be2496f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d47266c82ab0ca238b908179103d0b8be2496f78"}]},"branch":"refs/heads/master"},"4a9c3ff101fe26253a634997d149aa8e4a3dfe60":{"kind":"REWORK","_number":4,"created":"2025-05-09 00:02:34.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/20/946220/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/20/946220/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/20/946220/4"}}},"commit":{"parents":[{"commit":"d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb","subject":"Merge \"Parse json strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-04-03 00:02:41.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-05-09 00:02:33.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I181969ef8251a4c08e111ab824b26e4da1214ed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4a9c3ff101fe26253a634997d149aa8e4a3dfe60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4a9c3ff101fe26253a634997d149aa8e4a3dfe60"}]},"branch":"refs/heads/master"},"9d3df637902ec402e405c7e8586b5b0debabc987":{"kind":"REWORK","_number":5,"created":"2025-06-13 00:02:29.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/20/946220/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/20/946220/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/20/946220/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/20/946220/5"}}},"commit":{"parents":[{"commit":"d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb","subject":"Merge \"Parse json strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-04-03 00:02:41.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-13 00:02:28.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I181969ef8251a4c08e111ab824b26e4da1214ed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
