)]}'
{"id":"x%2Fpackstack~948173","triplet_id":"x%2Fpackstack~stable%2F2025.1~I13f380fb210664d71bcf6ece5de06c8b7063c55b","project":"x/packstack","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2025-06-13 03:49:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I13f380fb210664d71bcf6ece5de06c8b7063c55b","subject":"Switch to stable/2025.1 branch","status":"MERGED","created":"2025-04-25 07:36:28.000000000","updated":"2025-06-13 03:49:38.000000000","submitted":"2025-06-13 03:49:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948173","meta_rev_id":"003be9aea7376c92e6e380651293dac7a6541871","_number":948173,"virtual_id_number":948173,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-13 03:49:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-06-13 02:10:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-06-13 02:10:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-25 08:14:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-13 02:10:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"b0faf402ff361a510111d9f495645ead84626f8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2025-04-25 07:36:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48cf35a02940655bba0926b6d7ea3453c5412d60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 08:14:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95158c8dc39c4435b8f756450b8f07bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd17f216f29a4ef68446e2e325a9da65 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89e9cf4e3cb846d8885ac623979fb8a6 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cdc4ad4b2f2141d38344ef825e042000 : SUCCESS in 2m 55s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d2ab61a902b44e7a8c3f7bbfa3d1685e : SUCCESS in 21m 24s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/26a938ef7ca64b85bccefa3f97dfbc36 : SUCCESS in 32m 20s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/207a64d99a474dad9369bb49d0e01b6d : SUCCESS in 21m 20s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/792a6b14794e476eaf3a3d47a77dffdc : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4a074b9dd4a30127035905c3558c102739cfb7bf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-13 02:10:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b814bdd1f95671f250bdb523fd2cc415aebc369c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 02:12:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5c003b819749e580fd698f7ba62e7f068f16063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 03:49:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/825fab006d8b469cb696051086352b28\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/711ac9c31ad546f482e0f2dce58e9ae9 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5a668e9c55e436faa4a18bb2521e770 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9917c7756b5b4e60a3bc160ba782fa28 : SUCCESS in 3m 40s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/74a75a271c744c05a80103208b6f5b6f : SUCCESS in 27m 11s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cd3d6d4951e64e33a823f92b94cb8614 : SUCCESS in 40m 56s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/87c7b61611f44ea8a6489418964fe139 : SUCCESS in 51m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/740c46fbdc40475d96e3a50d58596d22 : SUCCESS in 48m 41s","accounts_in_message":[],"_revision_number":1},{"id":"003be9aea7376c92e6e380651293dac7a6541871","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 03:49:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd0e445527e524a4d34082ebb1ec2ec8d8b1b645","revisions":{"cd0e445527e524a4d34082ebb1ec2ec8d8b1b645":{"kind":"REWORK","_number":1,"created":"2025-04-25 07:36:28.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/73/948173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/73/948173/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/73/948173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/73/948173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/73/948173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/73/948173/1"}}},"commit":{"parents":[{"commit":"d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb","subject":"Merge \"Parse json strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d1dedfcf3daaf98d2cb3da3830778d0dc4d0fcbb"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2025-04-25 07:36:13.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2025-04-25 07:36:13.000000000","tz":120},"subject":"Switch to stable/2025.1 branch","message":"Switch to stable/2025.1 branch\n\nChange-Id: I13f380fb210664d71bcf6ece5de06c8b7063c55b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cd0e445527e524a4d34082ebb1ec2ec8d8b1b645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cd0e445527e524a4d34082ebb1ec2ec8d8b1b645"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
