)]}'
{"id":"x%2Fpackstack~953043","triplet_id":"x%2Fpackstack~master~I1063f0a240bfd336a3bd2a080db77c9cf1d4e493","project":"x/packstack","branch":"master","topic":"packstack-constraints","attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-09-19 00:02:15.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"37373":{"account":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"last_update":"2025-09-19 01:02:32.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1063f0a240bfd336a3bd2a080db77c9cf1d4e493","subject":"Updated from Packstack modules constraints","status":"NEW","created":"2025-06-22 00:02:22.000000000","updated":"2026-09-15 00:29:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9e4fce230a79ee4cf2b29baacc21f3031b20d539","_number":953043,"virtual_id_number":953043,"owner":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-15 00:29:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-22 00:41:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-23 06:49:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"403b70008d1e8b6884cae399d037bd7b0b312377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2025-06-22 00:02:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa1683fae7770d040e6dcbe12b70be5228d56cf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 00:41:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/710307914bb2419280db465eb9cd17da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8434ffdbf7fa4dc783ee53803d6ec90e : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4086155fb1ce47898a6cf7b1691f66fa : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5422a39904d4144898f528791174dab : SUCCESS in 3m 08s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e93b47e821414167926c21f68d58a593 : SUCCESS in 38m 25s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/497a0489c7714be99cf0b5f97dc04412 : SUCCESS in 30m 34s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/cb638ebae235435ba6caa50c32b6574c : SUCCESS in 31m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/15094998ba04499fac67838dee4bb885 : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c16dbc0f5ea094dd4e978135ae5c0d22db1b2cd0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-06-23 06:49:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9662aea4b54aa760a7bf14ffd3c87a127dc4ea1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2025-09-19 00:02:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"54523a97bf5c6ec3cbf08909a1bb4b59a6ddabd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 01:02:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e0bd46836d749748301ffb112dbea1a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9102251943c4c698221bd981b8ff278 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25ffc1ea1b5c44fa877d925d5016aa36 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4de63efbf4164d18bbf4f3d826b1629e : SUCCESS in 3m 02s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7cbe56cd612d434d9f8248d1f7cd2e03 : FAILURE in 56m 29s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0bc461dfb21b4d32a5dd2e9a80184a76 : FAILURE in 53m 30s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/de3b4d2cfb0f4af18afd33f7ff1ead02 : FAILURE in 49m 20s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3306f871f3e942b8985ba9dd761eea67 : FAILURE in 47m 28s","accounts_in_message":[],"_revision_number":2},{"id":"ea9e409663c7cd194d3b702acd6e855436425e06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2025-10-16 00:02:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5a4393288210566d5ffc56ab49cf1965bce9291d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 01:02:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3891a9d6ccf64ae493875de38405e2e5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e41703a226be4bca9803fc7d106bcfdc : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8305f49ca647492cb773d8bade8e9730 : SUCCESS in 2m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c70c910a0a64af7ab57e42f1ac9625a : SUCCESS in 3m 27s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/17b2838e29404f46805dd64349b6d075 : SUCCESS in 55m 31s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d7bfb372fb1e47cab4768a0ad900f237 : SUCCESS in 29m 19s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ac054d8ec5ae4276942061beb4dc25e4 : SUCCESS in 49m 46s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b1f813b9d6ee4af1954e38abffb5b29e : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b5b123f257aa7c0d9021f58a011e4720979f530c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2026-02-19 00:02:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8ee4a67b49b8e674c54cdfa5c306431031c5c2e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 01:01:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/713f67036b3a46acaea3ef5a58e30819\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380aec6a09d14431bde1845a4f8918e1 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59c0c11f3b0d4d91aeaced718266d4f2 : FAILURE in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b07379445dca487eb7cbe6683eef8e6d : SUCCESS in 3m 15s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/577683b142ab4a2f8e03de4694141f10 : SUCCESS in 30m 01s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4f1ce1916f444d50a2fdd8eadd8912d9 : SUCCESS in 42m 26s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6923286f06d14ccb8acb4a4355085e27 : SUCCESS in 53m 59s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/dfd9192c23dc41a9938f72c87fb783d5 : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":4},{"id":"d8566e4b7526e887f9dbe2c58868fe67dbd6dcfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2026-06-05 00:04:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"55de3f6fe6804300ad3e83e54a750270603caa92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 00:30:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29c0e9c56fed498b93822776ac22e809\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a9ff969b494ad8a12dc15c22f1114d : SUCCESS in 1m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdd2055920414d71af9d3e9092b3d2cf : FAILURE in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b32fd1f2c0554a8e9f5e02a07346843b : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ef07e84540f49bdb400dad11037a448 : SUCCESS in 3m 14s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/55eb9bfa84df468e95a6bea6faabbc64 : FAILURE in 7m 26s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7f336f9ba86a479b8e877721e112d5a6 : FAILURE in 8m 44s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7835c2edcfbb41369de821e441f5e8bd : FAILURE in 8m 40s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0c5cb10eb54847fd8298e592a2447fea : FAILURE in 12m 52s","accounts_in_message":[],"_revision_number":5},{"id":"be178d7cc6e9a956de3eb4e67ac8c37218c6d642","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2026-06-09 00:02:23.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"8f7407fb5593d51569f3970af65bd8590b60f903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 00:21:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d5383e056854858952438f43f8ff36b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943ec2d7231d454a9da11d6fd407c5bb : SUCCESS in 1m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e17462133be2477e9c2c0043314aa9f4 : FAILURE in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e24f9ff629849a39222efa14903c9e2 : SUCCESS in 2m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae4e30e2170241419409e1243408f818 : SUCCESS in 2m 06s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/82251a6f1e5f4ae8abb56fdac8e00a87 : FAILURE in 11m 18s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/22c772bbcb7e431ebb5cae61e627e140 : FAILURE in 7m 40s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b17a626db8df498398666af0a5f18429 : FAILURE in 7m 49s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9320bb074ad54174bb8e3b18a8326b64 : FAILURE in 11m 42s","accounts_in_message":[],"_revision_number":6},{"id":"86e271ace20355c6a69363a929af45aab84a8940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2026-06-12 00:03:00.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"b8856f169ebd85fdce8dfc233c4f5b05f0440618","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2026-07-03 00:03:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0694f32fb0134e0226d8a40da8f1ef360bd323e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 01:27:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa145dcc238542ec865b47d492332608\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96bfa664e6924b74b1d21a8c412d1e60 : SUCCESS in 2m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5350ea1a131849bdaf972c48dbdc2b95 : FAILURE in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5ce37b0de9144e79cefbf7af9268871 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34be7d00121a4b9cb0c37e4964d08bbc : SUCCESS in 2m 56s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/258b8ccff8d94dda8901946c5afa8070 : FAILURE in 11m 10s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8aa17b494632465fa1ec5fc301d8926b : FAILURE in 7m 55s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c9f1c46c302e42f282bb8c20e06d4f6b : FAILURE in 8m 28s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1e5bf0a5730144b8aa455ddf6ae9a7dd : FAILURE in 19m 21s","accounts_in_message":[],"_revision_number":8},{"id":"bd767ccae70ff184696c69ce88a9332f09242f40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2026-07-06 00:02:47.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"74a4da3cd232471383377b3dc4eb27398297edd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 00:26:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/754dd1c7ce3845ad847ddf0276a45d57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f977602671d944ea8bf8943215b70edc : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/071a5ae019c745c28dd14689737b589d : FAILURE in 2m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/130a3caa3c5e453f9d9eec8fafbb2cce : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5034811110c14d1496dfb135f2279a74 : SUCCESS in 2m 54s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d7031ce8876e40b2b7fe77181f9f4a7a : FAILURE in 18m 28s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/574ad964c9934043960e25b6222ec979 : FAILURE in 16m 33s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a4367ccb9d984907b969a8b002f2c1ac : FAILURE in 8m 48s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c6293d18bd1f4d23ae6cf55cd0e622d3 : FAILURE in 9m 54s","accounts_in_message":[],"_revision_number":9},{"id":"f5bd322b75c36f5c85c640a014bd810df5353487","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2026-08-15 00:04:09.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"91d53083609ae9ea6607895d8f8fdcab776bd5a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 00:32:45.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de7772d96f114ca0ba33adb74ed93c07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7581c44317684ec0a63f932776989463 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff5bbf3ff39f4a9ea57caad205b7bc54 : FAILURE in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e4689486d2e4d56891e1ca9cc841437 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/633930ccbbb24fcf87a18a47caf762ee : SUCCESS in 3m 07s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d9fb13d9cacb4b02ae20cc693449a82b : FAILURE in 8m 42s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/496aeb3ee2934039ad93ab01e03a2453 : FAILURE in 19m 31s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c1f23b0fbcde4407a5ec610fb9135c3e : FAILURE in 9m 02s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9b92861eb2c04c3296d28322b0bf8eca : FAILURE in 23m 19s","accounts_in_message":[],"_revision_number":10},{"id":"b78757518193523b9e9e37cbc3fc247e0b4a4dfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2026-09-04 00:04:34.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"813d512424d477b506edac16cfe3a2dd0dbd8c8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 00:20:10.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6839dd980074516b429f1ebc6d158b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d742de7f68d44c88337f766c5e5dc52 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bbe839d8f0644a9889d3ac2780c6697 : FAILURE in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3fdf8b370a7942e5993b45654fc3ab9c : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebfb0ce7d74a4cd9b5673442925208b1 : SUCCESS in 2m 07s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ee9ee60475ce405cb0c66f754b8710ac : FAILURE in 8m 25s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/61bb29bd60e94dc7abd5f4489ed7df2f : FAILURE in 8m 44s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1f5fef0395ce4c32b0c0a7289398f872 : FAILURE in 8m 26s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/83a49066bede4735bf4be7ffa4344f6f : FAILURE in 9m 16s","accounts_in_message":[],"_revision_number":11},{"id":"1802492228d0a96d1e6eb6f8a4db3653dcae0afc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2026-09-12 00:03:20.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"cc27c76eeb535ae3cd1f2d620a4c0ad4143770ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 00:23:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdc7d32989874c61a241c20a4bf3de7d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d60a4bdb837c4fa89ada3ba49d3db9c5 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9091f0cc16044579727f5a24be42657 : FAILURE in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/24b0e1e5285b4a799b99b7aca7014364 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c5fdd736e70476faeb7440b5c241b5f : SUCCESS in 3m 01s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/df8cbbaa63564a43bb9193b0aa569da8 : FAILURE in 9m 10s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1e5efb49b9c544bbbf04b532e879c21a : FAILURE in 9m 24s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5f0bf1d500694479a57657914c3d3a1e : FAILURE in 12m 19s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cb02cb183ef84677accb1ce51c137288 : FAILURE in 11m 27s","accounts_in_message":[],"_revision_number":12},{"id":"b83ae69efe1992e322ca2104d852760b7e90cd97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"date":"2026-09-15 00:03:30.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"9e4fce230a79ee4cf2b29baacc21f3031b20d539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 00:29:25.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5b4478f7c3e42aca97be8e4f0e6e5d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0734230314a446c97722d2188ae66af : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35da0aedd1824c03a2f8f5c25d1d3733 : FAILURE in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86e9c8732cd1418a8a66201987353203 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3249ff32301446c8d6a1a17fdc7b002 : SUCCESS in 3m 30s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0e703f1d67904011b9a83e113906c904 : FAILURE in 17m 47s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d0bf42661dc0496ca83402337897bb8f : FAILURE in 7m 53s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/4c3bb556ba404b5c95e07a0215d01d5e : FAILURE in 18m 30s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bbafcbe4bb5a4b8383dde6fc900dec8e : FAILURE in 18m 06s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"ff8567085a7336096eb06345852c6374cbea7195","revisions":{"d274b038403da95d1d8f106816ca96aac4648b01":{"kind":"REWORK","_number":1,"created":"2025-06-22 00:02:22.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/1"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d274b038403da95d1d8f106816ca96aac4648b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d274b038403da95d1d8f106816ca96aac4648b01"}]},"branch":"refs/heads/master"},"a61c20c20f4e891a6c3833f5108a28d1c80f99f9":{"kind":"REWORK","_number":2,"created":"2025-09-19 00:02:15.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/2"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-09-19 00:02:15.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a61c20c20f4e891a6c3833f5108a28d1c80f99f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a61c20c20f4e891a6c3833f5108a28d1c80f99f9"}]},"branch":"refs/heads/master"},"24f953fefdad352ec28e8d4e1996a4f403f542ab":{"kind":"REWORK","_number":3,"created":"2025-10-16 00:02:28.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/3"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-10-16 00:02:27.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/24f953fefdad352ec28e8d4e1996a4f403f542ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/24f953fefdad352ec28e8d4e1996a4f403f542ab"}]},"branch":"refs/heads/master"},"52ed18bff3ce841f1c7eac8348f44fea5aa58237":{"kind":"REWORK","_number":4,"created":"2026-02-19 00:02:19.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/4"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2026-02-19 00:02:18.000000000","tz":-300},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/52ed18bff3ce841f1c7eac8348f44fea5aa58237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/52ed18bff3ce841f1c7eac8348f44fea5aa58237"}]},"branch":"refs/heads/master"},"f1aa2aa7fdcd8b7ae104dfcea01f6fc2422c5fd5":{"kind":"REWORK","_number":5,"created":"2026-06-05 00:04:11.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/5"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2026-06-05 00:04:10.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f1aa2aa7fdcd8b7ae104dfcea01f6fc2422c5fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f1aa2aa7fdcd8b7ae104dfcea01f6fc2422c5fd5"}]},"branch":"refs/heads/master"},"fe2ed6dc0da3afe9e4418db267b3c5ef9addb687":{"kind":"REWORK","_number":6,"created":"2026-06-09 00:02:23.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/6","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/6"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2026-06-09 00:02:22.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fe2ed6dc0da3afe9e4418db267b3c5ef9addb687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fe2ed6dc0da3afe9e4418db267b3c5ef9addb687"}]},"branch":"refs/heads/master"},"c56d14a68338cda7abcc3c338cd621f308c0e0f5":{"kind":"REWORK","_number":7,"created":"2026-06-12 00:03:00.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/7","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/7"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2026-06-12 00:02:59.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c56d14a68338cda7abcc3c338cd621f308c0e0f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c56d14a68338cda7abcc3c338cd621f308c0e0f5"}]},"branch":"refs/heads/master"},"fabd99fc26951e0cd46d103bffd45cf0565b427e":{"kind":"REWORK","_number":8,"created":"2026-07-03 00:03:08.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/8","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/8"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2026-07-03 00:03:07.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fabd99fc26951e0cd46d103bffd45cf0565b427e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fabd99fc26951e0cd46d103bffd45cf0565b427e"}]},"branch":"refs/heads/master"},"01aa8ee6d4a6cd24245fa020c481b0c56947d7df":{"kind":"REWORK","_number":9,"created":"2026-07-06 00:02:47.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/9","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/9"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2026-07-06 00:02:46.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/01aa8ee6d4a6cd24245fa020c481b0c56947d7df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/01aa8ee6d4a6cd24245fa020c481b0c56947d7df"}]},"branch":"refs/heads/master"},"33db9e220180d157b94e9fbc16a7565f94351741":{"kind":"REWORK","_number":10,"created":"2026-08-15 00:04:09.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/10","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/10"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2026-08-15 00:04:09.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/33db9e220180d157b94e9fbc16a7565f94351741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/33db9e220180d157b94e9fbc16a7565f94351741"}]},"branch":"refs/heads/master"},"fc2d23fe008022821bc6fa2e244bc7cba0c7c438":{"kind":"REWORK","_number":11,"created":"2026-09-04 00:04:34.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/11","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/11"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2026-09-04 00:04:33.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fc2d23fe008022821bc6fa2e244bc7cba0c7c438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fc2d23fe008022821bc6fa2e244bc7cba0c7c438"}]},"branch":"refs/heads/master"},"bc2aa022b837312bed5cebd82aeebc93b4afc7a7":{"kind":"REWORK","_number":12,"created":"2026-09-12 00:03:20.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/12","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/12"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2026-09-12 00:03:19.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bc2aa022b837312bed5cebd82aeebc93b4afc7a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bc2aa022b837312bed5cebd82aeebc93b4afc7a7"}]},"branch":"refs/heads/master"},"ff8567085a7336096eb06345852c6374cbea7195":{"kind":"REWORK","_number":13,"created":"2026-09-15 00:03:30.000000000","uploader":{"_account_id":37373,"name":"RDO Release Bot","email":"rhos-role-rdo-admin@redhat.com","username":"rdo-release-bot"},"ref":"refs/changes/43/953043/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/953043/13","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/953043/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/953043/13"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2025-06-22 00:02:21.000000000","tz":-240},"committer":{"name":"rdo-release-bot","email":"rhos-role-rdo-admin@redhat.com","date":"2026-09-15 00:03:30.000000000","tz":-240},"subject":"Updated from Packstack modules constraints","message":"Updated from Packstack modules constraints\n\nChange-Id: I1063f0a240bfd336a3bd2a080db77c9cf1d4e493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ff8567085a7336096eb06345852c6374cbea7195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ff8567085a7336096eb06345852c6374cbea7195"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
