)]}'
{"id":"x%2Fpackstack~963341","triplet_id":"x%2Fpackstack~master~I754ba9b54bb8fc662ebf07697264c07e13ac82b1","project":"x/packstack","branch":"master","topic":"define_placement_api_service_name","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2025-10-08 01:30:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I754ba9b54bb8fc662ebf07697264c07e13ac82b1","subject":"Define api_service_name for placement::api","status":"ABANDONED","created":"2025-10-07 22:35:36.000000000","updated":"2025-10-08 01:30:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"528e5f1afa1edb0d5aed4c2ea5337a773ec311ff","_number":963341,"virtual_id_number":963341,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-07 22:58:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":-1,"date":"2025-10-07 22:52:10.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-07 22:58:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b8ea163ea48a6ff2a0653ebbd59f968c6dd9359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-07 22:35:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d246ca4f20601734f9bef3f06f1c8f7e2fa02658","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-07 22:52:10.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4a6ff3bad0b6dc6cf9efcefca5ed44ed5c7e702f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 22:58:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88aa3c8471564403bf58144cd43d12c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68184ef092bd41f19a71bca931326353 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb46580d23214c8e95958e43acab9794 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bbfd0e578d2748fa82853a08b59b2eb6 : SUCCESS in 4m 48s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6719a7d870c847fe80b985cc4e8ac95f : FAILURE in 11m 18s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/681afd6b8bb34eb7a967296517a2cfc0 : FAILURE in 11m 50s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/88a5e816b1c142ceb282605f7225b472 : FAILURE in 16m 36s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f8b39cae0fb943fab13b453dd9ca261b : FAILURE in 18m 23s","accounts_in_message":[],"_revision_number":1},{"id":"528e5f1afa1edb0d5aed4c2ea5337a773ec311ff","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 01:30:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95baecd2b84712e53ef99d5421363d777aca3e1c","revisions":{"95baecd2b84712e53ef99d5421363d777aca3e1c":{"kind":"REWORK","_number":1,"created":"2025-10-07 22:35:36.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/41/963341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/41/963341/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/41/963341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/41/963341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/41/963341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/41/963341/1"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-07 22:33:26.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-07 22:33:26.000000000","tz":120},"subject":"Define api_service_name for placement::api","message":"Define api_service_name for placement::api\n\nProbably related to Change: Ie93f2181e291bbf019379e5171924eaef4aee1c5\n\n$placement::params::service_name is ``undef`` for RedHat distributions.\n\nChange-Id: I754ba9b54bb8fc662ebf07697264c07e13ac82b1\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/95baecd2b84712e53ef99d5421363d777aca3e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/95baecd2b84712e53ef99d5421363d777aca3e1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
