)]}'
{"id":"x%2Fpackstack~963343","triplet_id":"x%2Fpackstack~master~I033ca7b4e0f5ffcdeedceaf4f716e93dbad68577","project":"x/packstack","branch":"master","topic":"ssh-keyscan-quiet","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2025-10-14 18:56:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I033ca7b4e0f5ffcdeedceaf4f716e93dbad68577","subject":"ssh-keyscan use quiet mode","status":"ABANDONED","created":"2025-10-07 23:58:10.000000000","updated":"2025-10-14 18:56:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5ef097c2cd03900d0914cf8a2cea7e54307ebe66","_number":963343,"virtual_id_number":963343,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-08 00:24:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-08 00:24:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"065ddf1a5c0d97b28455d732079acf060f34a3f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-07 23:58:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e976733a66961f88bdb3eb5136e10e4fe6bdfae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 00:24:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88ca8f9b364648fdbfe40251648a4661\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5db3517c6f44173887996a8d81bb3c4 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e72fe3e919774f63ba9c6bfa99463238 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d444ba96a5604c9a8b79f6c95971bb15 : SUCCESS in 2m 44s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9b81c333d0a1495ba0c8d3d9238141d7 : FAILURE in 16m 52s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b7b74c0d92d847d2ad776b636f7fae90 : FAILURE in 17m 40s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e8a4287ee3c74e87879a4560a8f60ef1 : FAILURE in 16m 39s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3ce31ba49d124204ab255e39e7a203a9 : FAILURE in 20m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5ef097c2cd03900d0914cf8a2cea7e54307ebe66","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-14 18:56:26.000000000","message":"Abandoned\n\nIncluded in another change fixing CI: https://review.opendev.org/c/x/packstack/+/963345/16/packstack/plugins/nova_300.py","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d2397c965ba547894afd8e293689b0bd070679c","revisions":{"7d2397c965ba547894afd8e293689b0bd070679c":{"kind":"REWORK","_number":1,"created":"2025-10-07 23:58:10.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/43/963343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/963343/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/963343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/963343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/963343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/963343/1"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-07 23:43:27.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-07 23:57:59.000000000","tz":120},"subject":"ssh-keyscan use quiet mode","message":"ssh-keyscan use quiet mode\n\nWhen not running in quiet mode ssh-keyscan prints server\nhost name and banners in comments. e.g the output is;\n\n10.1.200.21 ssh-rsa AAAAB3NzaC....\n\nThe create_compute_manifest() will split the first line, and\nset 10.1.200.21:22 as the ``host_key_type`` which is incorrect.\n\nChange-Id: I033ca7b4e0f5ffcdeedceaf4f716e93dbad68577\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7d2397c965ba547894afd8e293689b0bd070679c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7d2397c965ba547894afd8e293689b0bd070679c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
