)]}'
{"id":"x%2Fpackstack~963345","triplet_id":"x%2Fpackstack~master~I8304673138de5620e66009d71df68992a729d6df","project":"x/packstack","branch":"master","topic":"fix-jobs","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2025-10-11 15:40:59.000000000","reason":"\u003cGERRIT_ACCOUNT_24245\u003e replied on the change","reason_account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-12 11:07:41.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I8304673138de5620e66009d71df68992a729d6df","subject":"Use apache to run neutron api","status":"MERGED","created":"2025-10-08 01:10:29.000000000","updated":"2025-10-14 18:55:07.000000000","submitted":"2025-10-12 12:01:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963345-fix-jobs","meta_rev_id":"8f33e56b19de0b27bb04184bdb4c6eca9b879cdf","_number":963345,"virtual_id_number":963345,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-12 12:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":1,"date":"2025-10-11 15:40:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2025-10-12 11:07:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2025-10-12 11:07:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-08 01:31:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-08 14:26:20.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"CC"},{"updated":"2025-10-11 11:45:23.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"}],"messages":[{"id":"4928a36ac34a6a1ed5148d1aba3d97f6310ebc96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-08 01:10:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"205302e1654ec40d3dc49e15d489be1ab844df29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 01:31:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9a048dff0ee4e5ea26f0085c2b5585f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29d658f71d244679889b7debbc975fc5 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32ca5ae4ec7f4ef0bc90779d162fa556 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b94abc0189d74dc1a44cd06f05050b6a : SUCCESS in 3m 25s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2e98c3c4fef94eb389290d87b3ce5ceb : FAILURE in 17m 06s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/840f187611c845e3aa9338172a1fe555 : FAILURE in 7m 08s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7a6f916fcfc541e38b2e0da6a983b72e : FAILURE in 15m 22s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fce06c3c95f1463cb563d9e327484517 : FAILURE in 11m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0e8f3252040322cc0b69cbea4e60869dca0492a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-08 02:02:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7f05a1a5549672a940e2645c463a023694291dbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 03:02:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5eb43c527c9419193767e0954e4b39a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f90f8a21dcb84aada3d1c30f43898071 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8da341ab16314128b533df494df4fc8c : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/baac633000ad44f1b9d859b412091ec1 : SUCCESS in 3m 34s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6638897947ba4e9db0fbd146780763c1 : FAILURE in 18m 12s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/aad687e8dc7e4e51b2c84832ed03b7a8 : FAILURE in 8m 25s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/146aa38b44b541c2be4dd97c3301becb : FAILURE in 16m 39s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b83eb8cb57ab4754b260ab3939310bb1 : FAILURE in 19m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4c98e42beeb8f4a51de5f2391ec8abcaecc121c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-08 04:03:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5ea0755e94e4e0a4513f64ae283c91f76484fe82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 04:24:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17651cd43cdf42d590f26dd850e4d5fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b96aac8510e7482eba54452c99cbcca1 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c339c4ecf58b46f595cf353cd6e480aa : SUCCESS in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4aa7bb01ba0d4391a3e89522203937f4 : SUCCESS in 5m 13s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/88c1d991ca0e42cda9b77e96a31d2dcc : FAILURE in 7m 42s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2853a5a1eec4454f910f88981e77f8d1 : FAILURE in 7m 13s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7e05df9b544843b1a3748338e355c08d : FAILURE in 17m 05s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8a8da6a35f724150b761ee948bb6a6f6 : FAILURE in 8m 46s","accounts_in_message":[],"_revision_number":3},{"id":"26ac452be9e294281588e7987487356ec84ea76b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 07:53:43.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f1d3fd13049da73f936e07e07546a20d7e9c60da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 08:06:20.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"12ded712525a65708d7b7f7c288fd233f71b91e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 08:26:53.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"cfaa00bdb78c7cfc5cbeb2b5bf5ae10f12ce6864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 09:06:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4919869663f648f780fc342d11a9cef0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/300f31488c87408f9e814fee7f5fa2c5 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63c174308c2147a29168f4ab2a7043dd : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06cb6774e5434e86a1b40ea8c3d23fc5 : SUCCESS in 3m 13s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/37682472923543d48733434f91bbd501 : FAILURE in 33m 19s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bca31f7af63f49d28f16ce8325bc3ac7 : FAILURE in 33m 46s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f3d1ea9248f64b76a4da8c38739abca6 : FAILURE in 23m 37s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9db6f3f324eb44fe9935e4cb75b1356d : FAILURE in 15m 12s","accounts_in_message":[],"_revision_number":6},{"id":"b1a8bec640ca87c54e80ce77a859bcf61fbfe510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-08 12:53:58.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"0f3c9167de4965294c11de5dd39a595f6b70cfcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 13:09:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f05e9d28839745d687a2c128bb9864e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3590fe4b10a341b1b3f933bc2e898f4e : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6d7f845692c473f9ad5c7fc45adbb88 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c7b9a7f584140a29c75991d2c88cf50 : SUCCESS in 3m 32s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/40f66783adb340b0b545df3e71f06ead : FAILURE in 11m 43s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e2d54cf6624f4b22b30636877c2b1140 : FAILURE in 9m 57s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/02713005d01e423cb51906cc6b423893 : FAILURE in 9m 03s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e20f90ee2b214be9af38ce9e4b9d52b6 : FAILURE in 9m 01s","accounts_in_message":[],"_revision_number":7},{"id":"1c80f8176791de1ccab7843dde73e39100a29ae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-08 13:30:18.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"58fccbe2168d83f7a6a5278d128f0aeca34fc60d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 13:56:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dfd06bd57ba4afaab5a4f875c5fe368\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50ddaf07544c415e8aed551de0db8051 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afa1f3fe00fe4aaf97c72f02d9e68847 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b106537e910644bf97c381f6f06ddbb5 : SUCCESS in 3m 24s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/236af37a403744fd8bbc9409448bee2b : FAILURE in 16m 11s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/55776ff617d2448193ef2ea403a1060b : FAILURE in 9m 49s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/73abe5888c284caabe9c8ff28b12db18 : FAILURE in 10m 01s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f54a8b45994746ce86435600c0673d96 : FAILURE in 10m 35s","accounts_in_message":[],"_revision_number":8},{"id":"71d2f00147ba80ba591bf07c58e8b09887e75383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-08 13:58:31.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"c0a8dd007b3f11b2ee8ca47b32cb1b04a3ad573a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 14:21:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69697441cd1d4a95b4c903639c11a0cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/112c76225102401dad5e5c8cb4720146 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ffaebbf1ac34a349bf1f314c6bb1bb6 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/475acf994b104a4c82fd251323e032f9 : SUCCESS in 3m 50s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1cbbffad6c804d959766a2ccd80287e4 : FAILURE in 18m 24s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/65b2acff57b445c6b2faab4ddccd4543 : FAILURE in 9m 50s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7a002ba301924939946c5e0da49a21e8 : FAILURE in 9m 59s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b5055184139941efb12cfd9a318034fe : FAILURE in 12m 43s","accounts_in_message":[],"_revision_number":9},{"id":"485e42561ecf45ee6a98b17b08fb0a2f2ac26e7d","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 14:26:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4a2f715cbcbceaae05900e4b2f8b8493d6657056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 18:03:18.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"d9d2ee7c961198f816231e61f7ad7bbd7a35f948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 18:23:41.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0794ba65136f66931cda5397f974c3c3f04c5f8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 19:22:44.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9dbc4c5fef374cd4b4b0f92d725f5976\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33602af61313496986dad53254131bbe : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0afcd6a27f5f4da5a58e0ea95a3fb872 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df31d4ba75fa4c1eaf5629b0cd909373 : SUCCESS in 3m 48s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d7c96f3b317a40dfab1d5d7d61f90e72 : FAILURE in 21m 35s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bdba8e110a2e466ea3d7844223763f3d : SUCCESS in 34m 07s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d2726ca3cc0e45d2a3ff0658b26b42af : FAILURE in 37m 53s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d950960b69584391a08179882fc3acfd : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":11},{"id":"324ba12ecd200e8809b380ed623942a10a418c39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 20:40:29.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"e4aed6a6b3dc65f6d650cb4f29fb10befa62497d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 21:21:13.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"cfef351b8fe30c7c1e051bb601635bebf66f3ac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 22:11:02.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"80be463a18d58c14d0acc68812683d3ee5d8059b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 22:44:11.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/puppet-neutron/+/963469 is needed.","accounts_in_message":[],"_revision_number":14},{"id":"06ea03955253a1dd7192bc44b2fd0c573eed4e14","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 22:44:32.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"59dbd8d09cca8ae2566209ee1754c5d251ae8880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 23:24:28.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/puppet-neutron/+/963469 is needed.","accounts_in_message":[],"_revision_number":14},{"id":"e9dc79fff41cc814d4b85e55dc0028fd2e0deaa1","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-08 23:24:46.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d1707e50daa75c0706c156629a91ba1974df2ed5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 00:24:37.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2aa80cab76624a91870ec0220c7ce180\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1649c8e45e324cb08eb279bae6852aad : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d35af1ed30d45dfa13d199934e10808 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc7a5e240c0c4355bd0924aebde34e67 : SUCCESS in 3m 39s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/07dc547729dd4b1bbda30cb78c1cb20a : FAILURE in 35m 30s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/fa7f30535226401ebe80d82825581b03 : SUCCESS in 54m 21s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/64278e756ebc4a6f897e2ea4212be550 : FAILURE in 24m 26s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e524df464cba4232944ed215103d6e16 : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":14},{"id":"3097e1d02d9b4df7b256e4ef7fd85220301b2937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-09 07:42:24.000000000","message":"Patch Set 15: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"daad6b7568ae7c8d5706b69050a38b25ffa34306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 08:40:27.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2594595bebab479a9a81b6d43e636996\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3e77d0dbbc44410af335c51cc2dd4c0 : SUCCESS in 2m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3cfbad8d66145c289fb2c2af01752d4 : SUCCESS in 2m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2a3b944aa3444d6b1e8de8127619af5 : SUCCESS in 2m 22s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/09837cb84ac64de388abcd602138c6e8 : FAILURE in 18m 23s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/85c1e47f06c94ab180c284acd54a6aff : SUCCESS in 35m 02s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f5285995ba224e068b9d254f5c753c91 : FAILURE in 16m 09s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/83c0971792b64f06bb44c6649839b81a : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":15},{"id":"6c9674b134cb89ba48cd629432952ed6c34285d7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-11 10:05:27.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"020fb44c278f09236759f171c1390b79b9f3b37f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 10:59:48.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d2d8d27cc964cacb5e407e5344c754e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e9a92518777430a9674abff87560c50 : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9aeea8eef739474b822c20d87378ccc1 : SUCCESS in 2m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4ee4dd6a53141d690777410e646cd9a : SUCCESS in 3m 48s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/71177fdddc974130a9921ceb53420e4b : SUCCESS in 49m 25s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0d795d8517b1449086c1d2d9adcd642b : SUCCESS in 28m 39s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2c06e655e468481ab6821629c4ff4aa5 : SUCCESS in 49m 45s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e1bc938f93774d14afe9a2aaa8e313e7 : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":15},{"id":"b6c4628d2f8f5b077f26235574331d4640c7754f","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-11 11:45:23.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"9ae4e4e680c1aa204a09fceeebec99a53eb96260","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-11 14:13:50.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"5d7579c9daabec4416933b9f71fd255ce5db13ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-11 14:14:25.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"0a77b73d4297c4b6ee0b8b4fb7bcbb586a95e66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 15:11:06.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e40fe92c222446e3877c3b6f188ab19a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f427b655f7446ea3ebd4985584e602 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4120fbf5d9c941cabfb230b9e2cfc026 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8d8958212564595b17f068f9bd42f40 : SUCCESS in 3m 07s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5f3da6f0645b499ab1f41a25114452d0 : SUCCESS in 52m 43s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/baa8822e621b489798bd57b3c3d32e00 : SUCCESS in 48m 53s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6e058f5a611f42cc9dcde6b12e344ea2 : SUCCESS in 47m 38s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3551dfa17f1c492fa103cd4ddbe8359c : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":16},{"id":"56b1540392afb007e91f625fcbb399080ea9d08e","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-11 15:40:59.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"6f34015b94c7b05fc44c89a73ea5aa432ad37b2f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-12 11:07:41.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"3b0e4808890c0e4a249fa254e0f531e9d8974b31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-12 11:08:06.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"4a61c116ff064c3f4878e98416104f89a382ce11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-12 12:01:57.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6994c1def0749b293d76f17fcc04dfa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bd2a3b63bc8489bbd6ec232b5f4e1b3 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10da5c46c15041d597b28609b3e4f0d4 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ccea531347ad4685a98dc5691485066a : SUCCESS in 3m 22s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b088f19c03874e989663be45b26620e8 : SUCCESS in 46m 26s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7413172a57494262872769c1f97f0d3f : SUCCESS in 45m 50s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/bda831bcfc83426f9993b5baff16b5a7 : SUCCESS in 23m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/787d1ef3741a4e3ebff5d8b812a945ce : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":16},{"id":"ce1c65f1f388ac3c2c86b5452390baebf530ef48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-12 12:01:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"8f33e56b19de0b27bb04184bdb4c6eca9b879cdf","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-10-14 18:55:07.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"f581a3d98a79eb43f13ca0e981bead27c336f097","revisions":{"87d244ebc8d230e51acbf91cf3cd28409ed5cc02":{"kind":"REWORK","_number":1,"created":"2025-10-08 01:10:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/963345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/1"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:10:19.000000000","tz":540},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/87d244ebc8d230e51acbf91cf3cd28409ed5cc02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/87d244ebc8d230e51acbf91cf3cd28409ed5cc02"}]},"branch":"refs/heads/master"},"eddb98c145ee614ed9b3b3895b3718c0ffd54934":{"kind":"REWORK","_number":2,"created":"2025-10-08 02:02:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/963345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/2"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 02:02:31.000000000","tz":540},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eddb98c145ee614ed9b3b3895b3718c0ffd54934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eddb98c145ee614ed9b3b3895b3718c0ffd54934"}]},"branch":"refs/heads/master"},"80a0b6c7fc2b8289a41812e9e891b7ac7aa696ce":{"kind":"REWORK","_number":3,"created":"2025-10-08 04:03:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/963345/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/3"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 04:02:54.000000000","tz":540},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/80a0b6c7fc2b8289a41812e9e891b7ac7aa696ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/80a0b6c7fc2b8289a41812e9e891b7ac7aa696ce"}]},"branch":"refs/heads/master"},"e014d6a386d67b8e46598211fea1afe910584a40":{"kind":"REWORK","_number":4,"created":"2025-10-08 07:53:43.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/45/963345/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/4"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 07:53:43.000000000","tz":0},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e014d6a386d67b8e46598211fea1afe910584a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e014d6a386d67b8e46598211fea1afe910584a40"}]},"branch":"refs/heads/master"},"45417a2cc943862afc5c8237b38aab6fa7f6d236":{"kind":"REWORK","_number":5,"created":"2025-10-08 08:06:20.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/45/963345/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/5"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 08:06:20.000000000","tz":0},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/45417a2cc943862afc5c8237b38aab6fa7f6d236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/45417a2cc943862afc5c8237b38aab6fa7f6d236"}]},"branch":"refs/heads/master"},"cde230c4b8c91429cd7aa8acc93d6c674af713bf":{"kind":"REWORK","_number":6,"created":"2025-10-08 08:26:53.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/45/963345/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/6","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/6"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 08:26:53.000000000","tz":0},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cde230c4b8c91429cd7aa8acc93d6c674af713bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cde230c4b8c91429cd7aa8acc93d6c674af713bf"}]},"branch":"refs/heads/master"},"9df639855d643f9e858500ea9585c3bb6d9f9ef8":{"kind":"REWORK","_number":7,"created":"2025-10-08 12:53:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/963345/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/7","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/7"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 12:53:49.000000000","tz":540},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9df639855d643f9e858500ea9585c3bb6d9f9ef8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9df639855d643f9e858500ea9585c3bb6d9f9ef8"}]},"branch":"refs/heads/master"},"ec6368daa3645f76c134cb852621ddd33a6cd9da":{"kind":"REWORK","_number":8,"created":"2025-10-08 13:30:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/963345/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/8","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/8"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 13:30:10.000000000","tz":540},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ec6368daa3645f76c134cb852621ddd33a6cd9da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ec6368daa3645f76c134cb852621ddd33a6cd9da"}]},"branch":"refs/heads/master"},"75609cbd0e73169a735f51538e3566e636fa387d":{"kind":"REWORK","_number":9,"created":"2025-10-08 13:58:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/963345/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/9","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/9"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 13:58:23.000000000","tz":540},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/75609cbd0e73169a735f51538e3566e636fa387d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/75609cbd0e73169a735f51538e3566e636fa387d"}]},"branch":"refs/heads/master"},"5e174279fbb87087bac7b253c5b86329b781f942":{"kind":"REWORK","_number":10,"created":"2025-10-08 18:03:18.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/45/963345/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/10","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/10"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 18:02:53.000000000","tz":120},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5e174279fbb87087bac7b253c5b86329b781f942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5e174279fbb87087bac7b253c5b86329b781f942"}]},"branch":"refs/heads/master"},"00f5ac9ee50972bc0c2efee18d25a01436f953fe":{"kind":"REWORK","_number":11,"created":"2025-10-08 18:23:41.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/45/963345/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/11","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/11"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 18:22:44.000000000","tz":120},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nAlso, fixes some minor issues cause CI job failure.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/00f5ac9ee50972bc0c2efee18d25a01436f953fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/00f5ac9ee50972bc0c2efee18d25a01436f953fe"}]},"branch":"refs/heads/master"},"eb0ec327eb933f0a0669139664b6a47340718635":{"kind":"REWORK","_number":12,"created":"2025-10-08 20:40:29.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/45/963345/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/12","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/12"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 20:40:16.000000000","tz":120},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nAlso, fixes some minor issues cause CI job failure.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eb0ec327eb933f0a0669139664b6a47340718635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eb0ec327eb933f0a0669139664b6a47340718635"}]},"branch":"refs/heads/master"},"8ed0b86ce48d0fca3efd0ff318069474d0f11118":{"kind":"REWORK","_number":13,"created":"2025-10-08 21:21:13.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/45/963345/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/13","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/13"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 21:21:06.000000000","tz":120},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nAlso, fixes some minor issues cause CI job failure.\n\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8ed0b86ce48d0fca3efd0ff318069474d0f11118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8ed0b86ce48d0fca3efd0ff318069474d0f11118"}]},"branch":"refs/heads/master"},"ba709cf3a106530ab4ea3fdbd810e602b2ca36bc":{"kind":"REWORK","_number":14,"created":"2025-10-08 22:11:02.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/45/963345/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/14","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/14"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-10-08 22:10:41.000000000","tz":120},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nAlso, fixes some minor issues cause CI job failure.\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-neutron/+/963469\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ba709cf3a106530ab4ea3fdbd810e602b2ca36bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ba709cf3a106530ab4ea3fdbd810e602b2ca36bc"}]},"branch":"refs/heads/master"},"c6bf8cd3340a8dcbcc155b16c25c955d051c550c":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2025-10-09 07:42:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/963345/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/15","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/15"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-09 07:42:24.000000000","tz":0},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nAlso, fixes some minor issues cause CI job failure.\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-neutron/+/963364\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c6bf8cd3340a8dcbcc155b16c25c955d051c550c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c6bf8cd3340a8dcbcc155b16c25c955d051c550c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f581a3d98a79eb43f13ca0e981bead27c336f097":{"kind":"REWORK","_number":16,"created":"2025-10-11 14:13:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/963345/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/45/963345/16","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/45/963345/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/45/963345/16"}}},"commit":{"parents":[{"commit":"9d3df637902ec402e405c7e8586b5b0debabc987","subject":"Updated from Packstack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d3df637902ec402e405c7e8586b5b0debabc987"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-08 01:06:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-11 14:13:32.000000000","tz":540},"subject":"Use apache to run neutron api","message":"Use apache to run neutron api\n\nThe monolithic neutron-server service was removed due to removal of\neventlet library.\n\nAlso enable quiet mode in ssh-keyscan command to omit garbage records.\n\nCo-Authored-By: Harald Jensås \u003chjensas@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/puppet-neutron/+/963364\nChange-Id: I8304673138de5620e66009d71df68992a729d6df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f581a3d98a79eb43f13ca0e981bead27c336f097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f581a3d98a79eb43f13ca0e981bead27c336f097"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
