)]}'
{"id":"x%2Fpython-tuskarclient~123648","triplet_id":"x%2Fpython-tuskarclient~master~I987a7953b44ce27a93d0b29f3934a416bcdc93f2","project":"x/python-tuskarclient","branch":"master","topic":"bug/1368910","hashtags":[],"change_id":"I987a7953b44ce27a93d0b29f3934a416bcdc93f2","subject":"Stop using intersphinx","status":"MERGED","created":"2014-09-24 07:07:19.000000000","updated":"2014-09-25 19:57:25.000000000","submitted":"2014-09-25 19:57:24.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6cc425741eb041a11ae1107ee7de9958477bbce9","_number":123648,"virtual_id_number":123648,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2014-09-25 19:57:24.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"value":0,"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"recommended":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":1,"date":"2014-09-24 07:17:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2014-09-24 09:17:56.000000000","_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"value":2,"date":"2014-09-25 19:55:16.000000000","_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"value":1,"date":"2014-09-25 19:55:16.000000000","_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-09-24 07:17:55.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2014-09-24 09:17:56.000000000","updated_by":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"reviewer":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"state":"REVIEWER"},{"updated":"2014-09-25 19:55:16.000000000","updated_by":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"reviewer":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"state":"REVIEWER"},{"updated":"2014-09-25 19:57:24.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f39f7935f996eead3cd118aa588860bf270e74bd","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2014-09-24 07:07:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f654b2f33c77e4adaa42a9114e5495c05e3735e4","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2014-09-24 07:17:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a2e97cda8728717592d137d261a225ad922d3d32","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-24 07:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-python-tuskarclient-pep8 http://logs.openstack.org/48/123648/1/check/gate-python-tuskarclient-pep8/b60b272 : SUCCESS in 1m 07s\n- gate-python-tuskarclient-docs http://docs-draft.openstack.org/48/123648/1/check/gate-python-tuskarclient-docs/bbb1bac/doc/build/html/ : SUCCESS in 47s\n- gate-python-tuskarclient-python26 http://logs.openstack.org/48/123648/1/check/gate-python-tuskarclient-python26/c272c22 : SUCCESS in 58s\n- gate-python-tuskarclient-python27 http://logs.openstack.org/48/123648/1/check/gate-python-tuskarclient-python27/e8136bc : SUCCESS in 37s\n- gate-python-tuskarclient-python33 http://logs.openstack.org/48/123648/1/check/gate-python-tuskarclient-python33/653f2dc : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9a62b0e016a531d9dfa27affef1392c5233d18ed","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-24 08:36:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/48/123648/1/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/2608fd0 : SUCCESS in 49m 09s\n- check-tripleo-novabm-undercloud-precise-nonha http://logs.openstack.org/48/123648/1/check-tripleo/check-tripleo-novabm-undercloud-precise-nonha/4085b5a : SUCCESS in 47m 34s\n- check-tripleo-novabm-overcloud-precise-nonha http://logs.openstack.org/48/123648/1/check-tripleo/check-tripleo-novabm-overcloud-precise-nonha/b4cd621 : SUCCESS in 1h 21m 12s\n- check-tripleo-novabm-overcloud-f20-nonha http://logs.openstack.org/48/123648/1/check-tripleo/check-tripleo-novabm-overcloud-f20-nonha/5537e03 : SUCCESS in 1h 29m 15s","accounts_in_message":[],"_revision_number":1},{"id":"26fb44e2a68d46aa29a35a7419bf54c9dc1448e5","author":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"date":"2014-09-24 09:17:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d34e13313af4c3c23328897fa40568a2972bfc62","author":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"date":"2014-09-25 19:55:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a447e30c0d661b9e948577645192eae3a821068d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-25 19:55:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"4c82d9f718fdec6d7a35d762554fb330b0d81ae8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-25 19:57:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded.\n\n- gate-python-tuskarclient-docs http://docs-draft.openstack.org/48/123648/1/gate/gate-python-tuskarclient-docs/8b3c8c0/doc/build/html/ : SUCCESS in 1m 13s\n- gate-python-tuskarclient-pep8 http://logs.openstack.org/48/123648/1/gate/gate-python-tuskarclient-pep8/3e4ff40 : SUCCESS in 1m 03s\n- gate-python-tuskarclient-python26 http://logs.openstack.org/48/123648/1/gate/gate-python-tuskarclient-python26/cd279c6 : SUCCESS in 1m 11s\n- gate-python-tuskarclient-python27 http://logs.openstack.org/48/123648/1/gate/gate-python-tuskarclient-python27/782a2ea : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a26725e02b682ed7c64ddbf8d489a71ded6863b6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-25 19:57:25.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e0171b8ee630d17dc2f345a9daf6ad3fce2acf7","revisions":{"7e0171b8ee630d17dc2f345a9daf6ad3fce2acf7":{"kind":"REWORK","_number":1,"created":"2014-09-24 07:07:19.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/48/123648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/python-tuskarclient","ref":"refs/changes/48/123648/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/48/123648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/48/123648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/48/123648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/python-tuskarclient refs/changes/48/123648/1"}}},"commit":{"parents":[{"commit":"f2799bfc78f597f6f0fe37e4d68ea06beeac081f","subject":"Adds Templates class and use it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/f2799bfc78f597f6f0fe37e4d68ea06beeac081f"}]}],"author":{"name":"Christian Berendt","email":"berendt@b1-systems.de","date":"2014-09-24 07:07:08.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@b1-systems.de","date":"2014-09-24 07:07:08.000000000","tz":120},"subject":"Stop using intersphinx","message":"Stop using intersphinx\n\nRemove intersphinx from the docs build as it triggers network calls that\noccasionally fail, and we don\u0027t really use intersphinx (links other\nsphinx documents out on the internet)\n\nThis also removes the requirement for internet access during docs build.\n\nThis can cause docs jobs to fail if the project errors out on\nwarnings.\n\nChange-Id: I987a7953b44ce27a93d0b29f3934a416bcdc93f2\nRelated-Bug: #1368910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/7e0171b8ee630d17dc2f345a9daf6ad3fce2acf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/7e0171b8ee630d17dc2f345a9daf6ad3fce2acf7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f2799bfc78f597f6f0fe37e4d68ea06beeac081f","is_merged_in_target_branch":true,"change_id":"I484dac7c41cb90256aa1dee3eb997a4183635591","change_number":120729,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
