)]}'
{"id":"x%2Fpython-tuskarclient~131230","triplet_id":"x%2Fpython-tuskarclient~master~I2fb3c5ba50c6ef02473993c5fa2251240fbcd577","project":"x/python-tuskarclient","branch":"master","hashtags":[],"change_id":"I2fb3c5ba50c6ef02473993c5fa2251240fbcd577","subject":"Sync oslo-incubator","status":"MERGED","created":"2014-10-27 18:03:35.000000000","updated":"2014-10-31 09:07:43.000000000","submitted":"2014-10-31 09:07:43.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8880b0fc3aa92d106c3f1674241df09858f15756","_number":131230,"virtual_id_number":131230,"owner":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},{"value":2,"date":"2014-10-31 09:07:43.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"recommended":{"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},"all":[{"value":1,"date":"2014-10-28 04:59:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2014-10-29 09:24:14.000000000","_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":2,"date":"2014-10-27 20:06:19.000000000","_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"all":[{"value":0,"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-10-31 09:07:06.000000000","_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},{"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-10-27 20:06:19.000000000","updated_by":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"reviewer":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"state":"REVIEWER"},{"updated":"2014-10-28 04:59:38.000000000","updated_by":{"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},"reviewer":{"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},"state":"REVIEWER"},{"updated":"2014-10-31 09:07:06.000000000","updated_by":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"reviewer":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"state":"REVIEWER"},{"updated":"2014-10-31 09:07:43.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"1cce7a9e343ac2544c2b749b1c3e9c6932dbf990","author":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"date":"2014-10-27 18:03:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09742ec80c4a34aee0a16a03c23b237ee22af481","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-27 18:37:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-python-tuskarclient-pep8 http://logs.openstack.org/30/131230/1/check/gate-python-tuskarclient-pep8/d531d52 : SUCCESS in 30s\n- gate-python-tuskarclient-docs http://docs-draft.openstack.org/30/131230/1/check/gate-python-tuskarclient-docs/68657bc/doc/build/html/ : SUCCESS in 37s\n- gate-python-tuskarclient-python27 http://logs.openstack.org/30/131230/1/check/gate-python-tuskarclient-python27/5af83dd : SUCCESS in 34s\n- gate-python-tuskarclient-python26 http://logs.openstack.org/30/131230/1/check/gate-python-tuskarclient-python26/44544e2 : SUCCESS in 42s\n- gate-python-tuskarclient-python33 http://logs.openstack.org/30/131230/1/check/gate-python-tuskarclient-python33/7355532 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e7d200c56fce98da70d44e4b349819499d491c9d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-27 18:50:40.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/30/131230/1/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/7cf6ce6 : FAILURE in 35m 32s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/30/131230/1/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/a6d3d3e : FAILURE in 46m 56s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/30/131230/1/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/a1b2822 : FAILURE in 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9ff45109402207bc29bc0fffe5f9937f9c539ad6","author":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"date":"2014-10-27 20:06:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11f50ac3f59f1ebe7e7177da585ec21b2a300f43","author":{"_account_id":12385,"name":"Motohiro/Yuanying Otsuka","email":"yuanying@fraction.jp","username":"yuanying"},"date":"2014-10-28 04:59:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"39245cb1bf7ea1921f0af62ae1677047f3ef41dc","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2014-10-29 09:24:14.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"4e0221d830dea6d208b2f3566010ca26b7657425","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 09:26:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- gate-python-tuskarclient-pep8 http://logs.openstack.org/30/131230/1/check/gate-python-tuskarclient-pep8/052cdfc : SUCCESS in 25s\n- gate-python-tuskarclient-docs http://docs-draft.openstack.org/30/131230/1/check/gate-python-tuskarclient-docs/6f9e77b/doc/build/html/ : SUCCESS in 46s\n- gate-python-tuskarclient-python27 http://logs.openstack.org/30/131230/1/check/gate-python-tuskarclient-python27/4cab531 : SUCCESS in 28s\n- gate-python-tuskarclient-python26 http://logs.openstack.org/30/131230/1/check/gate-python-tuskarclient-python26/6c705f0 : SUCCESS in 37s\n- gate-python-tuskarclient-python33 http://logs.openstack.org/30/131230/1/check/gate-python-tuskarclient-python33/e44a6a6 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ec0c3a7daa255eabccb3d56136a920799e6fc387","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-29 11:02:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/30/131230/1/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/0f74ae8 : SUCCESS in 50m 59s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/30/131230/1/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/c3e308c : SUCCESS in 1h 37m 46s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/30/131230/1/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/52a21df : SUCCESS in 1h 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c4dc67598cb83110f3d5dd9ca4579723b3609c21","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-31 09:07:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"9ec7230fddad7263ca0cfbd610f7f1f2dd13b38a","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2014-10-31 09:07:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ecd81f43a940f6bfcd63a16f767b30a467615786","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-31 09:07:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-python-tuskarclient-docs http://docs-draft.openstack.org/30/131230/1/gate/gate-python-tuskarclient-docs/8698447/doc/build/html/ : SUCCESS in 24s\n- gate-python-tuskarclient-pep8 http://logs.openstack.org/30/131230/1/gate/gate-python-tuskarclient-pep8/23fd515 : SUCCESS in 24s\n- gate-python-tuskarclient-python27 http://logs.openstack.org/30/131230/1/gate/gate-python-tuskarclient-python27/d4faebf : SUCCESS in 27s\n- gate-python-tuskarclient-python26 http://logs.openstack.org/30/131230/1/gate/gate-python-tuskarclient-python26/d872b63 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1},{"id":"8880b0fc3aa92d106c3f1674241df09858f15756","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-31 09:07:43.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80d1fde7b8f082b43dd1ccf199a968e68ea0f00e","revisions":{"80d1fde7b8f082b43dd1ccf199a968e68ea0f00e":{"kind":"REWORK","_number":1,"created":"2014-10-27 18:03:35.000000000","uploader":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"ref":"refs/changes/30/131230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/python-tuskarclient","ref":"refs/changes/30/131230/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/30/131230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/30/131230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/30/131230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/python-tuskarclient refs/changes/30/131230/1"}}},"commit":{"parents":[{"commit":"71ac3c6c5e80a48848078e60d6fae7b168b52445","subject":"Add --version option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/71ac3c6c5e80a48848078e60d6fae7b168b52445"}]}],"author":{"name":"Steven Hardy","email":"shardy@redhat.com","date":"2014-10-27 17:41:33.000000000","tz":0},"committer":{"name":"Steven Hardy","email":"shardy@redhat.com","date":"2014-10-27 17:46:55.000000000","tz":0},"subject":"Sync oslo-incubator","message":"Sync oslo-incubator\n\nSync oslo-incubator to 5ab5bd3, this involves removing the reference\nto the deprecated (now removed in favor of oslo.utils) importutils.\n\nChange-Id: I2fb3c5ba50c6ef02473993c5fa2251240fbcd577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/80d1fde7b8f082b43dd1ccf199a968e68ea0f00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/80d1fde7b8f082b43dd1ccf199a968e68ea0f00e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"71ac3c6c5e80a48848078e60d6fae7b168b52445","is_merged_in_target_branch":true,"change_id":"Ia8a19602802a1da1029f8e575dafcafcca87388a","change_number":131211,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
