)]}'
{"id":"x%2Fpython-tuskarclient~256825","triplet_id":"x%2Fpython-tuskarclient~master~I3f3bea86572089f1cce36c3588d8d75a32af783a","project":"x/python-tuskarclient","branch":"master","hashtags":[],"change_id":"I3f3bea86572089f1cce36c3588d8d75a32af783a","subject":"Deprecated tox -downloadcache option removed","status":"ABANDONED","created":"2015-12-11 22:31:37.000000000","updated":"2016-01-22 15:27:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"400eea733a8ed77ecc64b53739441708b7826fe8","_number":256825,"virtual_id_number":256825,"owner":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":-1,"date":"2016-01-22 14:52:39.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":14867,"name":"Jaivish Kothari(janonymous)","email":"janonymous.codevulture@gmail.com","username":"Janonymous"},{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":2,"date":"2015-12-14 12:17:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-12-21 13:36:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14867,"name":"Jaivish Kothari(janonymous)","email":"janonymous.codevulture@gmail.com","username":"Janonymous"},{"value":1,"date":"2015-12-12 01:15:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":-1,"date":"2016-01-22 11:37:29.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":-1,"date":"2015-12-22 21:29:50.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":14867,"name":"Jaivish Kothari(janonymous)","email":"janonymous.codevulture@gmail.com","username":"Janonymous"},{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":14867,"name":"Jaivish Kothari(janonymous)","email":"janonymous.codevulture@gmail.com","username":"Janonymous"},{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-12-12 01:15:49.000000000","updated_by":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"reviewer":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"state":"REVIEWER"},{"updated":"2015-12-21 13:36:04.000000000","updated_by":{"_account_id":14867,"name":"Jaivish Kothari(janonymous)","email":"janonymous.codevulture@gmail.com","username":"Janonymous"},"reviewer":{"_account_id":14867,"name":"Jaivish Kothari(janonymous)","email":"janonymous.codevulture@gmail.com","username":"Janonymous"},"state":"REVIEWER"},{"updated":"2015-12-22 21:29:50.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2016-01-22 11:37:29.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2016-01-22 14:52:39.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"665f2148de6321a1f4e9a1d7be87e75b902b2d6e","author":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"date":"2015-12-11 22:31:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7984e9fa1411c8bae4fd444feb2a32b563de9dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-11 23:02:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-python-tuskarclient-pep8 http://logs.openstack.org/25/256825/1/check/gate-python-tuskarclient-pep8/cf0d35c/ : SUCCESS in 56s\n- gate-python-tuskarclient-docs http://docs-draft.openstack.org/25/256825/1/check/gate-python-tuskarclient-docs/3a7c66f//doc/build/html/ : SUCCESS in 1m 03s\n- gate-python-tuskarclient-python27 http://logs.openstack.org/25/256825/1/check/gate-python-tuskarclient-python27/64da436/ : SUCCESS in 56s\n- gate-python-tuskarclient-python34 http://logs.openstack.org/25/256825/1/check/gate-python-tuskarclient-python34/7710a77/ : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1},{"id":"96032b0695e33708695223cc7d74926a6f05fd5a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-11 23:46:00.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-f22-ceph http://logs.openstack.org/25/256825/1/check-tripleo/gate-tripleo-ci-f22-ceph/ab1f451/ : FAILURE in 7m 12s\n- gate-tripleo-ci-f22-ha http://logs.openstack.org/25/256825/1/check-tripleo/gate-tripleo-ci-f22-ha/0c8fd5e/ : FAILURE in 7m 46s\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/25/256825/1/check-tripleo/gate-tripleo-ci-f22-nonha/a7f1a03/ : FAILURE in 6m 56s\n- gate-tripleo-ci-f22-containers http://logs.openstack.org/25/256825/1/check-tripleo/gate-tripleo-ci-f22-containers/d99d0d7/ : FAILURE in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b876228b17e5641df44139d2be0e602f75b9767","author":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"date":"2015-12-12 01:15:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"24061e7038861d2b16698a1ee8205a8fa328d981","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2015-12-14 12:17:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"681d7417935e558d6a1a1ca6af653126c9070da1","author":{"_account_id":14867,"name":"Jaivish Kothari(janonymous)","email":"janonymous.codevulture@gmail.com","username":"Janonymous"},"date":"2015-12-21 13:36:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f3225e1c50d226d24ca3681b5e36868bfbd6347f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2015-12-22 21:29:50.000000000","message":"Patch Set 1: Workflow-1\n\nThis project is deprecated and slated for removal soon.  It\u0027s no longer built as part of the CI run so it\u0027s not possible for it to pass the tripleo-ci tests.","accounts_in_message":[],"_revision_number":1},{"id":"f52c904d24d98bb0bd40a65244c8ddcb1d128517","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2016-01-22 11:37:29.000000000","message":"Patch Set 1: Workflow-1\n\nThe code base is being retired. https://review.openstack.org/270855","accounts_in_message":[],"_revision_number":1},{"id":"325464ec45e9d06cbd6d9135a1c4c74b35be3ad4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-22 14:52:39.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"400eea733a8ed77ecc64b53739441708b7826fe8","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2016-01-22 15:27:56.000000000","message":"Abandoned\n\nAs Dougal notes, this has been retired so it can\u0027t merge anymore.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa8927a3470d190e633762e72173cfcbb8bfc06b","revisions":{"aa8927a3470d190e633762e72173cfcbb8bfc06b":{"kind":"REWORK","_number":1,"created":"2015-12-11 22:31:37.000000000","uploader":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"ref":"refs/changes/25/256825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/python-tuskarclient","ref":"refs/changes/25/256825/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/25/256825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/25/256825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/25/256825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/python-tuskarclient refs/changes/25/256825/1"}}},"commit":{"parents":[{"commit":"e77d2a1b6613847c951b447f3814400f46651e49","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/e77d2a1b6613847c951b447f3814400f46651e49"}]}],"author":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2015-12-11 22:11:59.000000000","tz":60},"committer":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2015-12-11 22:31:34.000000000","tz":60},"subject":"Deprecated tox -downloadcache option removed","message":"Deprecated tox -downloadcache option removed\n\nCaching is enabled by default from pip version 6.0\n\nMore info:\nhttps://testrun.org/tox/latest/config.html#confval-downloadcache\u003dpath\nhttps://pip.pypa.io/en/stable/reference/pip_install/#caching\n\nChange-Id: I3f3bea86572089f1cce36c3588d8d75a32af783a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/aa8927a3470d190e633762e72173cfcbb8bfc06b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/aa8927a3470d190e633762e72173cfcbb8bfc06b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e77d2a1b6613847c951b447f3814400f46651e49","is_merged_in_target_branch":true,"change_id":"I753562f07fc648726081f1cab5f0ec2dbd182849","change_number":194031,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
