)]}'
{"id":"x%2Fpython-tuskarclient~59163","triplet_id":"x%2Fpython-tuskarclient~master~I71b73998fe22aaf11cb39d26b428cbfe567c0398","project":"x/python-tuskarclient","branch":"master","topic":"fix_wheels","hashtags":[],"change_id":"I71b73998fe22aaf11cb39d26b428cbfe567c0398","subject":"Revert \"Support building wheels (PEP-427)\"","status":"MERGED","created":"2013-11-29 11:51:51.000000000","updated":"2013-11-29 15:41:33.000000000","submitted":"2013-11-29 15:41:33.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f443afabf5477d0b88c8dc6d38bfb61df8e57e1","_number":59163,"virtual_id_number":59163,"owner":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-11-29 15:41:33.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"value":0,"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"recommended":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-11-29 12:23:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":2,"date":"2013-11-29 15:40:43.000000000","_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"value":2,"date":"2013-11-29 11:53:52.000000000","_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-11-29 12:23:16.000000000","_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":1,"date":"2013-11-29 15:40:43.000000000","_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"value":0,"date":"2013-11-29 11:53:52.000000000","_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-11-29 11:53:52.000000000","updated_by":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"reviewer":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"state":"REVIEWER"},{"updated":"2013-11-29 12:23:16.000000000","updated_by":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"reviewer":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"state":"REVIEWER"},{"updated":"2013-11-29 15:40:43.000000000","updated_by":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"reviewer":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"state":"REVIEWER"},{"updated":"2013-11-29 15:41:33.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5609b0b8f6f5b17cbd83ae2598e456d13af95eda","author":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"date":"2013-11-29 11:53:52.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"030cec505407c2d799700cc7b152d13ae2a598c6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-29 11:55:53.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- gate-python-tuskarclient-pep8 http://logs.openstack.org/63/59163/1/check/gate-python-tuskarclient-pep8/9ef7276 : SUCCESS in 31s\n- gate-python-tuskarclient-python26 http://logs.openstack.org/63/59163/1/check/gate-python-tuskarclient-python26/78cd2c0 : SUCCESS in 53s\n- gate-python-tuskarclient-python27 http://logs.openstack.org/63/59163/1/check/gate-python-tuskarclient-python27/0d48c2d : SUCCESS in 43s\n- gate-python-tuskarclient-python33 http://logs.openstack.org/63/59163/1/check/gate-python-tuskarclient-python33/dd4bbd1 : FAILURE in 3m 57s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"6cad6980e17f88fddb78c91c633ebd877b54a4a1","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2013-11-29 12:23:16.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\nYeah we\u0027re close to having py3 but not there yet. I didn\u0027t realize we\u0027ll want to use wheels as distribution mechanism *that* soon :) In that case it makes sense to revert.","accounts_in_message":[],"_revision_number":1},{"id":"27623c203f6bd53b4857575ef5d377f7e7605e7a","author":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"date":"2013-11-29 15:40:43.000000000","message":"Patch Set 1: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":1},{"id":"722e45c2a9686b7eecff375380fe6fae0f3d84f0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-29 15:40:44.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"e5fa5093172a8ca76ca78b51ee27bcc1a4c11632","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-29 15:41:33.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1},{"id":"774bdc08e1285fb080adbb10bd04e2f4aab9a5f7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-29 15:41:33.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- gate-python-tuskarclient-pep8 http://logs.openstack.org/63/59163/1/gate/gate-python-tuskarclient-pep8/6eb93a6 : SUCCESS in 39s\n- gate-python-tuskarclient-python26 http://logs.openstack.org/63/59163/1/gate/gate-python-tuskarclient-python26/04f0cc6 : SUCCESS in 45s\n- gate-python-tuskarclient-python27 http://logs.openstack.org/63/59163/1/gate/gate-python-tuskarclient-python27/53656c5 : SUCCESS in 31s\n","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"771b7f484f7f5e99e6b5d534dd3a36d2351d8d07","revisions":{"771b7f484f7f5e99e6b5d534dd3a36d2351d8d07":{"kind":"REWORK","_number":1,"created":"2013-11-29 11:51:51.000000000","uploader":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"ref":"refs/changes/63/59163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/python-tuskarclient","ref":"refs/changes/63/59163/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/63/59163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/63/59163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/python-tuskarclient refs/changes/63/59163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/python-tuskarclient refs/changes/63/59163/1"}}},"commit":{"parents":[{"commit":"f3d6ec51cf18b6d8f939f5bc0772638949ada032","subject":"Support building wheels (PEP-427)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/f3d6ec51cf18b6d8f939f5bc0772638949ada032"}]}],"author":{"name":"Roman Podoliaka","email":"rpodolyaka@mirantis.com","date":"2013-11-29 11:49:33.000000000","tz":120},"committer":{"name":"Roman Podoliaka","email":"rpodolyaka@mirantis.com","date":"2013-11-29 11:51:27.000000000","tz":120},"subject":"Revert \"Support building wheels (PEP-427)\"","message":"Revert \"Support building wheels (PEP-427)\"\n\nThis reverts commit f3d6ec51cf18b6d8f939f5bc0772638949ada032.\n\nThe patch being reverted should not have been merged, because\nsetting universal to True is used to claim in-place python 2/3\ncompatibility in Wheel packages.\n\nWe don\u0027t support python 3 yet (issues with keystone client), so we\nmust not claim the opposite here (as this would harm end users as\nsoon as we start uploading Wheel packages to PyPi).\n\nPlease see the following links for details:\n - http://pythonwheels.com/\n - http://lists.openstack.org/pipermail/openstack-dev/2013-November/020059.html\n\nChange-Id: I71b73998fe22aaf11cb39d26b428cbfe567c0398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/771b7f484f7f5e99e6b5d534dd3a36d2351d8d07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/python-tuskarclient/commit/771b7f484f7f5e99e6b5d534dd3a36d2351d8d07"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f3d6ec51cf18b6d8f939f5bc0772638949ada032","is_merged_in_target_branch":true,"change_id":"I076739d63d3f0bee229a7b61cbf20671421eb9c4","change_number":57160,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
