)]}'
{"id":"x%2Fsqlalchemy-migrate~577002","triplet_id":"x%2Fsqlalchemy-migrate~master~I88ab886f0cd21249c1e81f2592e2a0ca1102fce3","project":"x/sqlalchemy-migrate","branch":"master","topic":"universal_wheels","attention_set":{},"removed_from_attention_set":{"28456":{"account":{"_account_id":28456,"name":"Nicola Soranzo","email":"nicola.soranzo@gmail.com","username":"nsoranzo"},"last_update":"2024-09-25 10:35:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I88ab886f0cd21249c1e81f2592e2a0ca1102fce3","subject":"Build universal wheels","status":"ABANDONED","created":"2018-06-20 22:26:20.000000000","updated":"2024-09-25 10:35:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2649507ac6942286148a5e1d09a4104dee8b2ca5","_number":577002,"virtual_id_number":577002,"owner":{"_account_id":28456,"name":"Nicola Soranzo","email":"nicola.soranzo@gmail.com","username":"nsoranzo"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2023-06-29 10:26:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"all":[{"value":2,"date":"2023-06-29 09:39:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"all":[{"value":1,"date":"2023-06-29 09:39:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 20:17:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-29 09:39:05.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"}],"messages":[{"id":"312696f9d1ecc427d47802627e4c7558520c1a92","author":{"_account_id":28456,"name":"Nicola Soranzo","email":"nicola.soranzo@gmail.com","username":"nsoranzo"},"date":"2018-06-20 22:26:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f6e4eb9e42c40e52df86f73d675e5d2b06d17bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-20 23:56:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/02/577002/1/check/build-openstack-sphinx-docs/0ddb691/ : FAILURE in 4m 15s\n- openstack-tox-py35 http://logs.openstack.org/02/577002/1/check/openstack-tox-py35/e89dd8b/ : FAILURE in 5m 31s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/02/577002/1/check/openstack-tox-pep8/58d5437/ : SUCCESS in 4m 30s (non-voting)\n- legacy-sqlalchemy-migrate-devstack-dsvm http://logs.openstack.org/02/577002/1/check/legacy-sqlalchemy-migrate-devstack-dsvm/b2668d2/ : SUCCESS in 1h 25m 06s (non-voting)\n- openstack-tox-py27 http://logs.openstack.org/02/577002/1/check/openstack-tox-py27/1573b31/ : FAILURE in 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5eb79152f5c611f944e35cf9b54e57a2718d5132","author":{"_account_id":28456,"name":"Nicola Soranzo","email":"nicola.soranzo@gmail.com","username":"nsoranzo"},"date":"2019-01-18 11:51:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f4d1695e767eb478aee099502277e7e5e130902e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-18 13:11:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/02/577002/2/check/build-openstack-sphinx-docs/05626bf/html/ : SUCCESS in 5m 59s\n- openstack-tox-py35 http://logs.openstack.org/02/577002/2/check/openstack-tox-py35/0624f11/ : FAILURE in 10m 03s (non-voting)\n- test-release-openstack http://logs.openstack.org/02/577002/2/check/test-release-openstack/c892b0d/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/02/577002/2/check/openstack-tox-pep8/8c83f04/ : SUCCESS in 5m 50s (non-voting)\n- legacy-sqlalchemy-migrate-devstack-dsvm http://logs.openstack.org/02/577002/2/check/legacy-sqlalchemy-migrate-devstack-dsvm/3a284b9/ : SUCCESS in 1h 17m 15s (non-voting)\n- openstack-tox-py27 http://logs.openstack.org/02/577002/2/check/openstack-tox-py27/51da4d8/ : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":2},{"id":"6014b5871e93771c903fbbe8556ba8eac08bf4fb","author":{"_account_id":28456,"name":"Nicola Soranzo","email":"nicola.soranzo@gmail.com","username":"nsoranzo"},"date":"2020-04-17 18:33:13.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a84f67671256b22939c6c9437fb1e6479e7898ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 20:17:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aae31ef80273460bbce4687cebb5e299 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c7000ca16c449486acd9055964502d : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0595e8f72a3747c485ea8d91f39e7444 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69053af83545466585ab614db2ba78f0 : SUCCESS in 12m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d9cb62c170fd449b9507a27f2e7de2a3 : SUCCESS in 10m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/879afe7fb3894131b96e5a86fe092c04 : SUCCESS in 7m 35s\n- sqlalchemy-migrate-tox-py27sa07 https://zuul.opendev.org/t/openstack/build/aa3b5010eb4d45ccabd61ad74292a95b : SUCCESS in 12m 37s\n- sqlalchemy-migrate-devstack https://zuul.opendev.org/t/openstack/build/ad0e28cc7695479bbcfb1e11372c2e5e : SUCCESS in 1h 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1858d1acf6f705bb3f67ad4e54a6a671ac023675","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-06-29 09:39:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"390627efe2837c474b95d33c96f4604dabbcaa55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 09:39:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"48672412eea3e94956711e8bb8799dbbbf8de071","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 10:26:55.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80b030c42c9c4db2afa06fb9dbf72c65\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc41bf04a2864fa8a1c04f0d4a66faab : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d381d25ce5f34afaaa58dbf4482f209f : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dad464cef454f9b9d3793f588cac2cb : FAILURE in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67148e13b22245caa2149f1cc6d96a0f : SUCCESS in 13m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/11a820f69b5b41808cd5a80f523b2b44 : TIMED_OUT in 40m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/284b0506783c4e948e93953b5cb79e18 : FAILURE in 3m 49s\n- sqlalchemy-migrate-tox-py27sa07 https://zuul.opendev.org/t/openstack/build/8aad4d22dc0741e6b8df1e26b5043823 : RETRY_LIMIT in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"2649507ac6942286148a5e1d09a4104dee8b2ca5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-25 10:35:02.000000000","message":"Abandoned\n\nRetiring repo","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"46edb1813aea42ea05c87ea630d1afcc56312b39","revisions":{"f36b5e4f7cd0b52871f1155e33382a3a3f56c882":{"kind":"REWORK","_number":1,"created":"2018-06-20 22:26:20.000000000","uploader":{"_account_id":28456,"name":"Nicola Soranzo","email":"nicola.soranzo@gmail.com","username":"nsoranzo"},"ref":"refs/changes/02/577002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/sqlalchemy-migrate","ref":"refs/changes/02/577002/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/1"}}},"commit":{"parents":[{"commit":"2b4d322b0677926f386ff4aa6edeb8df65f596d3","subject":"Add .eggs in .gitignore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/2b4d322b0677926f386ff4aa6edeb8df65f596d3"}]}],"author":{"name":"Nicola Soranzo","email":"nicola.soranzo@earlham.ac.uk","date":"2018-06-20 22:23:36.000000000","tz":60},"committer":{"name":"Nicola Soranzo","email":"nicola.soranzo@earlham.ac.uk","date":"2018-06-20 22:23:39.000000000","tz":60},"subject":"Build universal wheels","message":"Build universal wheels\n\nSee https://packaging.python.org/guides/distributing-packages-using-setuptools/#universal-wheels\n\nChange-Id: I88ab886f0cd21249c1e81f2592e2a0ca1102fce3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/f36b5e4f7cd0b52871f1155e33382a3a3f56c882"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/f36b5e4f7cd0b52871f1155e33382a3a3f56c882"}]},"branch":"refs/heads/master"},"4d3bfac9831f62d7f5d217fcb55355e4a2ce106d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-01-18 11:51:17.000000000","uploader":{"_account_id":28456,"name":"Nicola Soranzo","email":"nicola.soranzo@gmail.com","username":"nsoranzo"},"ref":"refs/changes/02/577002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/sqlalchemy-migrate","ref":"refs/changes/02/577002/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/2"}}},"commit":{"parents":[{"commit":"8fd7226f186427d8f005c00ca155322f2f72078b","subject":"Merge \"Add .eggs in .gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/8fd7226f186427d8f005c00ca155322f2f72078b"}]}],"author":{"name":"Nicola Soranzo","email":"nicola.soranzo@earlham.ac.uk","date":"2018-06-20 22:23:36.000000000","tz":60},"committer":{"name":"Nicola Soranzo","email":"nsoranzo@tiscali.it","date":"2019-01-18 11:51:17.000000000","tz":0},"subject":"Build universal wheels","message":"Build universal wheels\n\nSee https://packaging.python.org/guides/distributing-packages-using-setuptools/#universal-wheels\n\nChange-Id: I88ab886f0cd21249c1e81f2592e2a0ca1102fce3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/4d3bfac9831f62d7f5d217fcb55355e4a2ce106d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/4d3bfac9831f62d7f5d217fcb55355e4a2ce106d"}]},"branch":"refs/heads/master"},"46edb1813aea42ea05c87ea630d1afcc56312b39":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-17 18:33:13.000000000","uploader":{"_account_id":28456,"name":"Nicola Soranzo","email":"nicola.soranzo@gmail.com","username":"nsoranzo"},"ref":"refs/changes/02/577002/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/sqlalchemy-migrate","ref":"refs/changes/02/577002/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/sqlalchemy-migrate refs/changes/02/577002/3"}}},"commit":{"parents":[{"commit":"5d1f322542cd8eb42381612765be4ed9ca8105ec","subject":"remove inspect.getargspec deprecation warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/5d1f322542cd8eb42381612765be4ed9ca8105ec"}]}],"author":{"name":"Nicola Soranzo","email":"nicola.soranzo@earlham.ac.uk","date":"2018-06-20 22:23:36.000000000","tz":60},"committer":{"name":"Nicola Soranzo","email":"nicola.soranzo@gmail.com","date":"2020-04-17 18:33:13.000000000","tz":0},"subject":"Build universal wheels","message":"Build universal wheels\n\nSee https://packaging.python.org/guides/distributing-packages-using-setuptools/#universal-wheels\n\nChange-Id: I88ab886f0cd21249c1e81f2592e2a0ca1102fce3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/46edb1813aea42ea05c87ea630d1afcc56312b39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/46edb1813aea42ea05c87ea630d1afcc56312b39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Verified\u003dMIN"],"failing_atoms":["label:Verified\u003dMAX"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
