)]}'
{"id":"x%2Fsqlalchemy-migrate~670113","triplet_id":"x%2Fsqlalchemy-migrate~master~I4b7bccdb7cebd95753e8cd931846213f400aa652","project":"x/sqlalchemy-migrate","branch":"master","topic":"bindep","hashtags":[],"change_id":"I4b7bccdb7cebd95753e8cd931846213f400aa652","subject":"Add bindep support","status":"MERGED","created":"2019-07-10 16:16:06.000000000","updated":"2019-07-11 00:21:49.000000000","submitted":"2019-07-11 00:21:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670113-1562804509097-c3c832e5","meta_rev_id":"3bd4064d3a9de4be6661a5ff537312a288a37000","_number":670113,"virtual_id_number":670113,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":2,"date":"2019-07-11 00:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-07-10 16:17:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-07-11 00:03:37.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-07-11 00:03:37.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2019-07-10 18:16:28.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 16:17:44.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-10 18:16:28.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2019-07-11 00:21:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f658387f7461dad6677156cf0409c9c987a6d9a5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-10 16:16:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df29ee7da602e6f334b166c8e66efd0e0b743b1f","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-10 16:17:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09472e6a9b57d5aaca0b2b023f347ca269965d8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:06:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/13/670113/1/check/openstack-tox-docs/14ab6c1/html/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/13/670113/1/check/openstack-tox-pep8/87eee1b/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/13/670113/1/check/openstack-tox-py27/806ca7a/ : SUCCESS in 10m 44s\n- openstack-tox-py35 http://logs.openstack.org/13/670113/1/check/openstack-tox-py35/42ae75b/ : FAILURE in 6m 44s (non-voting)\n- sqlalchemy-migrate-tox-py27sa07 http://logs.openstack.org/13/670113/1/check/sqlalchemy-migrate-tox-py27sa07/dace3c4/ : SUCCESS in 11m 16s\n- sqlalchemy-migrate-devstack http://logs.openstack.org/13/670113/1/check/sqlalchemy-migrate-devstack/8f39f00/ : SUCCESS in 1h 42m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b315b40111c126b1cc6096d08b3e0f9c29792fe0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-11 00:03:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSelf-approving to unblock CI jobs and get the rest of the previously approved series merged.","accounts_in_message":[],"_revision_number":1},{"id":"68936ee720b882dbae3ac146886f440472872b6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 00:03:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21ecdff79fabb75172671dfd6ee8279a8310fe5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 00:21:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/13/670113/1/gate/openstack-tox-docs/5e75b8b/html/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/13/670113/1/gate/openstack-tox-pep8/4374607/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/13/670113/1/gate/openstack-tox-py27/ced8023/ : SUCCESS in 11m 22s\n- sqlalchemy-migrate-tox-py27sa07 http://logs.openstack.org/13/670113/1/gate/sqlalchemy-migrate-tox-py27sa07/e06b486/ : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6028b167adb157e5c1d35c371cf4e15248d6828d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 00:21:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ee28fe67b5c99e15db450b7f6c069dd66594a94","revisions":{"3ee28fe67b5c99e15db450b7f6c069dd66594a94":{"kind":"REWORK","_number":1,"created":"2019-07-10 16:16:06.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/13/670113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/sqlalchemy-migrate","ref":"refs/changes/13/670113/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/13/670113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/13/670113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/sqlalchemy-migrate refs/changes/13/670113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/sqlalchemy-migrate refs/changes/13/670113/1"}}},"commit":{"parents":[{"commit":"078ffc62dae3a93898a4e7ad5ba014087543af6f","subject":"Remove py26 tox targets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/078ffc62dae3a93898a4e7ad5ba014087543af6f"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-07-10 16:14:20.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-07-10 16:14:20.000000000","tz":-240},"subject":"Add bindep support","message":"Add bindep support\n\nThis adds a minimal bindep.txt to install mysql and postgresql\ndependencies (copied from nova\u0027s bindep.txt). This is needed\nnow to get python jobs (pep8,py27,etc) working in the gate.\n\nChange-Id: I4b7bccdb7cebd95753e8cd931846213f400aa652\nCloses-Bug: #1836087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/3ee28fe67b5c99e15db450b7f6c069dd66594a94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/sqlalchemy-migrate/commit/3ee28fe67b5c99e15db450b7f6c069dd66594a94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
