)]}'
{"id":"x%2Fstackalytics~841905","triplet_id":"x%2Fstackalytics~master~Icaaa7687131387653353fb378baa39420166aedf","project":"x/stackalytics","branch":"master","attention_set":{},"removed_from_attention_set":{"9569":{"account":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"last_update":"2023-01-15 23:23:30.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-01-15 23:23:30.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-01-15 23:23:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icaaa7687131387653353fb378baa39420166aedf","subject":"Change City Network company name to Cleura","status":"MERGED","created":"2022-05-16 14:29:57.000000000","updated":"2023-01-15 23:23:30.000000000","submitted":"2023-01-15 23:23:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841905","meta_rev_id":"75e5f331a3788fbba6f1aa7b0cb3169876ec898e","_number":841905,"virtual_id_number":841905,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-15 23:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"recommended":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-29 17:06:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2023-01-15 23:00:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2023-01-15 23:00:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-16 14:30:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"CC"},{"updated":"2022-05-16 14:38:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-17 06:07:01.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2022-06-22 08:23:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":9569,"name":"Sergey Nikitin","email":"nikitinserv@gmail.com","username":"snikitin"},"state":"REVIEWER"},{"updated":"2022-06-22 08:23:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-01-15 23:00:44.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"}],"messages":[{"id":"21f290a2aba48b4215e8ffc36b473abda9267d92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-16 14:29:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4e1cf1279a8dbe41c72d0ffad2554cfcef10757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 14:38:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3d2d77533a4ca29d13da58d00f9e88 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d44c54ddb24740fab715993d301cf4b2 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9540ad4bee714ea18db6affb623f213f : FAILURE in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6c599d5642d4f389e24ff401658bd76 : FAILURE in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/393eb31ff0134b1db79e2b76ddb92649 : FAILURE in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd0ccd7da84702dc877299b599f9b849c88da805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-16 14:43:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3bc060c8093ff5c89243b2da9836357239c77f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 14:54:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e30fda1cc3a4f47bb6b6b27afb4c767 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e2faf8ada904e718898c9877539e9a2 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38b42fdbad7c4f7e97ad16dd8b93024b : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f41399486d2c4aeea24d11168f124b34 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf166b212d614705bce25b9d9c1a6734 : SUCCESS in 3m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4580659d0fce9024f259f6140a6dce080dd65d1f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2022-05-17 06:07:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"175f2692260bcb8ff582920850fc50ab43747554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-29 17:06:08.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"beaeef544d81972d85fc2ab0553d9ce9aa945688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 17:26:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17a1b5d06844482eb4a77f48615dd745\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142a4442496a4091b46f6f875cbeec7f : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a029768e88d84b28962ac5508780fe4a : SUCCESS in 2m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2f3ff9818a246218befa178ee4a972c : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1704d526e9df489c9adb3391a06a95c3 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c65551e2fed3420589a4e8bbc57e262d : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"953fec8ff8ebc5ce0f78abb08264ac2350c020f4","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2023-01-15 23:00:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a8de3617220a51e5b35011616b0225849a14f5c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 23:01:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dd6712f6ff2983b3bc1169332b6956ad217d4470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 23:23:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1844b325a644626840d281339727fee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0deabbd4d08542c188b050e502daed93 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d58d5745304961b44d667147aa3dd3 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79b15bb9f3c14715b18bbad4304d5736 : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d60fb823d41b4a8199d3137940cca52c : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":3},{"id":"75e5f331a3788fbba6f1aa7b0cb3169876ec898e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 23:23:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5374a3bb1531df5e8c8e8846d643df4cbbbc8942","revisions":{"13a31da6cce49e4e1be9e2fddedfdd7acf1a5101":{"kind":"REWORK","_number":1,"created":"2022-05-16 14:29:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/841905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/stackalytics","ref":"refs/changes/05/841905/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/stackalytics refs/changes/05/841905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/stackalytics refs/changes/05/841905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/stackalytics refs/changes/05/841905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/stackalytics refs/changes/05/841905/1"}}},"commit":{"parents":[{"commit":"019e6683562e53b87d7b74122f1f9a31ae30b3f2","subject":"Add new user info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/019e6683562e53b87d7b74122f1f9a31ae30b3f2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-16 14:29:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-16 14:29:55.000000000","tz":120},"subject":"Change City Network company name to Cleura","message":"Change City Network company name to Cleura\n\nWith rebranding [1] of City Network to Cleura, we updating company\nname and adding old one as an alias.\nCleura domains are also being added to the list.\n\n[1] https://cleura.com/press-releases/city-network-becomes-cleura/\n\nChange-Id: Icaaa7687131387653353fb378baa39420166aedf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/13a31da6cce49e4e1be9e2fddedfdd7acf1a5101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/13a31da6cce49e4e1be9e2fddedfdd7acf1a5101"}]},"branch":"refs/heads/master"},"0202d5a9e79dda61c46b7e3687b3ed2c7cce4d54":{"kind":"REWORK","_number":2,"created":"2022-05-16 14:43:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/841905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/stackalytics","ref":"refs/changes/05/841905/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/stackalytics refs/changes/05/841905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/stackalytics refs/changes/05/841905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/stackalytics refs/changes/05/841905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/stackalytics refs/changes/05/841905/2"}}},"commit":{"parents":[{"commit":"019e6683562e53b87d7b74122f1f9a31ae30b3f2","subject":"Add new user info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/019e6683562e53b87d7b74122f1f9a31ae30b3f2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-16 14:29:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-16 14:43:25.000000000","tz":120},"subject":"Change City Network company name to Cleura","message":"Change City Network company name to Cleura\n\nWith rebranding [1] of City Network to Cleura, we updating company\nname and adding old one as an alias.\nCleura domains are also being added to the list.\n\n[1] https://cleura.com/press-releases/city-network-becomes-cleura/\n\nChange-Id: Icaaa7687131387653353fb378baa39420166aedf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/0202d5a9e79dda61c46b7e3687b3ed2c7cce4d54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/0202d5a9e79dda61c46b7e3687b3ed2c7cce4d54"}]},"branch":"refs/heads/master"},"5374a3bb1531df5e8c8e8846d643df4cbbbc8942":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-29 17:06:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/841905/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/stackalytics","ref":"refs/changes/05/841905/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/stackalytics refs/changes/05/841905/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/stackalytics refs/changes/05/841905/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/stackalytics refs/changes/05/841905/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/stackalytics refs/changes/05/841905/3"}}},"commit":{"parents":[{"commit":"ac4cf3538497d0b6af6be327770395bc544e4d9d","subject":"Add new OVH corp email for Arnaud Morin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/ac4cf3538497d0b6af6be327770395bc544e4d9d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-16 14:29:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-29 17:06:08.000000000","tz":0},"subject":"Change City Network company name to Cleura","message":"Change City Network company name to Cleura\n\nWith rebranding [1] of City Network to Cleura, we updating company\nname and adding old one as an alias.\nCleura domains are also being added to the list.\n\n[1] https://cleura.com/press-releases/city-network-becomes-cleura/\n\nChange-Id: Icaaa7687131387653353fb378baa39420166aedf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/5374a3bb1531df5e8c8e8846d643df4cbbbc8942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/5374a3bb1531df5e8c8e8846d643df4cbbbc8942"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}}]}],"submit_requirements":[]}
