)]}'
{"id":"x%2Fstackalytics~946012","triplet_id":"x%2Fstackalytics~master~I5526827f5cf4ae692c48bb2c3196e7aa5489f2a5","project":"x/stackalytics","branch":"master","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-04-15 14:00:16.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-04-15 14:00:16.000000000","reason":"Change was submitted"},"19519":{"account":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},"last_update":"2025-04-02 13:24:36.000000000","reason":"\u003cGERRIT_ACCOUNT_19519\u003e replied on the change","reason_account":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"}}},"hashtags":[],"change_id":"I5526827f5cf4ae692c48bb2c3196e7aa5489f2a5","subject":"Updating Ghanshyam Mann Contributor Affiliation","status":"MERGED","created":"2025-04-01 03:03:48.000000000","updated":"2025-04-15 14:00:16.000000000","submitted":"2025-04-15 14:00:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946012","meta_rev_id":"109f862c0ab7c4f30a9977f9541c7f40c9fb37a2","_number":946012,"virtual_id_number":946012,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-15 14:00:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-01 08:03:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-04-02 13:24:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-04-02 13:24:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-01 03:05:51.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},"state":"REVIEWER"},{"updated":"2025-04-01 03:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-01 07:54:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-04-01 07:54:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"e78613be1e7ea00de219fa82171819795d885c10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-04-01 03:03:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0e466a5918c525b899ddb18458992dfb4d65f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 03:22:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1c11ae1ab824899b58c602f2514a7a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70806390f04d4f81af239fed4b1e2d59 : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca029f3522944be0b8e82f1eb3e69574 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3158090667b94c278607f518e2734d2d : FAILURE in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a49197654868476a8afcf5c30945239d : FAILURE in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fee227bac5d2aa4821dc1515c1699ce916059e5d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-01 07:54:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"303019264d8ae5c9fd558f799fffc19295979a36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-01 07:54:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":2},{"id":"1a0ff4939cd59a6c8904247de75814913784df08","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-01 07:54:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a39fcdca5d85031b224004a869a26993a76670ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 07:59:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b3dff02df7745419cafe5d2f1935c4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/229709f4509249f99102e7f2c0d9bba7 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f943396f7a1415b82a8e91a5483823a : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac759530210f4bf9bdc2340a7b374cab : FAILURE in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bc1ab5f329943ee9e4f7d97979142ee : FAILURE in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"35eb1096dd8ea295badaf97b198b5f9d18ecc4c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-01 08:03:10.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8d7effbb0b5672d9fa907a95f952ce7e83067cf7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-01 08:03:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3296c2e7810fa8a13e570bbc6c8f17e50c7b387c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 08:12:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4752c1051224ab1b5c96107be657bb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e46583f0688472fafe270a2ef78bd1e : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84cb411a61bc4a1e9fa391203efc4d3f : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d386a8c48373464cbd2010bf8016d626 : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be548a1bd7bd451a839932f01cbb105c : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3},{"id":"1bc7fa4cf2ea597d35b8bdbeeba98d96d7bae4d8","author":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},"date":"2025-04-02 13:24:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0844bd7e951122040a56db8589860900001eb72e","author":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},"date":"2025-04-15 13:46:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"03accec7941b33a06b6e1ab8c6b55c5fe7546ee9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 13:50:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7209d6fabacd837ba248ad12f562dd8cf2d6bb52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 13:54:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/931314e4f6f948389e19781838ce719c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16141357367548d08c365b3883995381 : SUCCESS in 1m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d702cb99754a4a8f4612623b208e0d : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cff1d17e077f41bda45096fd409462e4 : SUCCESS in 1m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83321d777f3e49ff8ce90ac06357672d : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/501764d194374f3da10dd8fadf2b6fa2 : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0a7a425841cfc1e6f620efb014bb958e5de5edc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 14:00:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa06e7cab39c4e28bcaf47b14a89bbab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dcbe6a4225e44ca9c1d98128aa551dc : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f687062027834c33841d87406eb1120d : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a213c52cf1941eaa6437152b4653e89 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/269ce57865594f19b7c096b02b378ef2 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":3},{"id":"109f862c0ab7c4f30a9977f9541c7f40c9fb37a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 14:00:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9eab1896d2dfa19bd76e73932d5f399f57bbd44","revisions":{"3860728fe045ca6e511ebed1688fd2d14b62ed9f":{"kind":"REWORK","_number":1,"created":"2025-04-01 03:03:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/946012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/stackalytics","ref":"refs/changes/12/946012/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/stackalytics refs/changes/12/946012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/stackalytics refs/changes/12/946012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/stackalytics refs/changes/12/946012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/stackalytics refs/changes/12/946012/1"}}},"commit":{"parents":[{"commit":"f33de2186e4caa5c654a23f9d40f583524055bb3","subject":"Merge \"Fix outdated tox minversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/f33de2186e4caa5c654a23f9d40f583524055bb3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-04-01 03:03:09.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-04-01 03:03:47.000000000","tz":-420},"subject":"Updating Ghanshyam Mann Contributor Affiliation","message":"Updating Ghanshyam Mann Contributor Affiliation\n\nChange-Id: I5526827f5cf4ae692c48bb2c3196e7aa5489f2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/3860728fe045ca6e511ebed1688fd2d14b62ed9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/3860728fe045ca6e511ebed1688fd2d14b62ed9f"}]},"branch":"refs/heads/master"},"98a34a45f9d4ee01e0975b8cb139c63c5bff92c0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-01 07:54:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/946012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/stackalytics","ref":"refs/changes/12/946012/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/stackalytics refs/changes/12/946012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/stackalytics refs/changes/12/946012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/stackalytics refs/changes/12/946012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/stackalytics refs/changes/12/946012/2"}}},"commit":{"parents":[{"commit":"875d6ded6a06bbb4cc312e51f59e19158a4caef4","subject":"setup.cfg: Replace dashes by underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/875d6ded6a06bbb4cc312e51f59e19158a4caef4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-04-01 03:03:09.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2025-04-01 07:54:09.000000000","tz":0},"subject":"Updating Ghanshyam Mann Contributor Affiliation","message":"Updating Ghanshyam Mann Contributor Affiliation\n\nChange-Id: I5526827f5cf4ae692c48bb2c3196e7aa5489f2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/98a34a45f9d4ee01e0975b8cb139c63c5bff92c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/98a34a45f9d4ee01e0975b8cb139c63c5bff92c0"}]},"branch":"refs/heads/master","description":"Rebase"},"f9eab1896d2dfa19bd76e73932d5f399f57bbd44":{"kind":"REWORK","_number":3,"created":"2025-04-01 08:03:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/946012/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/stackalytics","ref":"refs/changes/12/946012/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/stackalytics refs/changes/12/946012/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/stackalytics refs/changes/12/946012/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/stackalytics refs/changes/12/946012/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/stackalytics refs/changes/12/946012/3"}}},"commit":{"parents":[{"commit":"875d6ded6a06bbb4cc312e51f59e19158a4caef4","subject":"setup.cfg: Replace dashes by underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/875d6ded6a06bbb4cc312e51f59e19158a4caef4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-04-01 03:03:09.000000000","tz":-420},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-04-01 08:03:10.000000000","tz":0},"subject":"Updating Ghanshyam Mann Contributor Affiliation","message":"Updating Ghanshyam Mann Contributor Affiliation\n\nChange-Id: I5526827f5cf4ae692c48bb2c3196e7aa5489f2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/f9eab1896d2dfa19bd76e73932d5f399f57bbd44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/stackalytics/commit/f9eab1896d2dfa19bd76e73932d5f399f57bbd44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
