)]}'
{"id":"zuul%2Fzuul~659708","triplet_id":"zuul%2Fzuul~master~If2e2324235a6608985168d606d95adeade36afe0","project":"zuul/zuul","branch":"master","topic":"tox-remote","hashtags":[],"change_id":"If2e2324235a6608985168d606d95adeade36afe0","subject":"zuul-tox-remote: use unique zuul_console service","status":"MERGED","created":"2019-05-17 03:28:46.000000000","updated":"2019-05-17 21:09:52.000000000","submitted":"2019-05-17 21:08:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659708-1558127303237-5d02b967","meta_rev_id":"b23a2bf727497c7cc63ed92b6aac39ef1ac9c0b9","_number":659708,"virtual_id_number":659708,"owner":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2019-05-17 19:19:03.000000000","_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":2,"date":"2019-05-17 21:08:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-05-17 12:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":2,"date":"2019-05-17 12:29:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":1,"date":"2019-05-17 12:29:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 12:13:46.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-05-17 12:29:22.000000000","updated_by":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"reviewer":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"state":"REVIEWER"},{"updated":"2019-05-17 19:19:03.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2019-05-17 21:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6176acf76e49e559e6804c8217306294f3f54c5","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2019-05-17 03:28:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f120d2e31eac3d69c42557faf67a389dbc618dd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 04:31:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- zuul-build-image http://logs.openstack.org/08/659708/1/check/zuul-build-image/cf392b1/ : SUCCESS in 29m 46s\n- tox-docs http://logs.openstack.org/08/659708/1/check/tox-docs/144f8cb/html/ : SUCCESS in 4m 02s\n- tox-pep8 http://logs.openstack.org/08/659708/1/check/tox-pep8/364a4ec/ : SUCCESS in 6m 08s\n- tox-py35 http://logs.openstack.org/08/659708/1/check/tox-py35/45b0f82/ : SUCCESS in 39m 03s\n- tox-py36 http://logs.openstack.org/08/659708/1/check/tox-py36/6b11420/ : SUCCESS in 37m 08s\n- nodejs-npm-run-lint http://logs.openstack.org/08/659708/1/check/nodejs-npm-run-lint/632aa7f/ : SUCCESS in 6m 11s\n- zuul-stream-functional-2.5 http://logs.openstack.org/08/659708/1/check/zuul-stream-functional-2.5/b1066cc/ : SUCCESS in 5m 57s\n- zuul-stream-functional-2.6 http://logs.openstack.org/08/659708/1/check/zuul-stream-functional-2.6/e3c92a4/ : SUCCESS in 6m 09s\n- zuul-stream-functional-2.7 http://logs.openstack.org/08/659708/1/check/zuul-stream-functional-2.7/ce47f5f/ : SUCCESS in 6m 28s\n- zuul-stream-functional-2.8 http://logs.openstack.org/08/659708/1/check/zuul-stream-functional-2.8/e2f7f08/ : SUCCESS in 7m 55s\n- zuul-tox-remote http://logs.openstack.org/08/659708/1/check/zuul-tox-remote/16ab32c/ : SUCCESS in 17m 50s\n- zuul-quick-start http://logs.openstack.org/08/659708/1/check/zuul-quick-start/a0b6524/ : SUCCESS in 7m 40s\n- nodepool-zuul-functional http://logs.openstack.org/08/659708/1/check/nodepool-zuul-functional/b8c134b/ : SUCCESS in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8e6ff549f9e1b8bfbd9ba87ea07f38942cd0a75","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2019-05-17 04:47:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8f4a918ce3cfbfc22cb48d42486cb1ded61b2bce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 05:25:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- zuul-build-image http://logs.openstack.org/08/659708/2/check/zuul-build-image/cd90adb/ : SUCCESS in 22m 54s\n- tox-docs http://logs.openstack.org/08/659708/2/check/tox-docs/a284883/html/ : SUCCESS in 4m 15s\n- tox-pep8 http://logs.openstack.org/08/659708/2/check/tox-pep8/9fb5e09/ : SUCCESS in 3m 57s\n- tox-py35 http://logs.openstack.org/08/659708/2/check/tox-py35/501303a/ : SUCCESS in 37m 17s\n- tox-py36 http://logs.openstack.org/08/659708/2/check/tox-py36/c08e1b3/ : SUCCESS in 35m 06s\n- nodejs-npm-run-lint http://logs.openstack.org/08/659708/2/check/nodejs-npm-run-lint/3e0d035/ : SUCCESS in 4m 52s\n- zuul-tox-remote http://logs.openstack.org/08/659708/2/check/zuul-tox-remote/2173334/ : TIMED_OUT in 30m 47s\n- zuul-quick-start http://logs.openstack.org/08/659708/2/check/zuul-quick-start/defe303/ : SUCCESS in 7m 14s\n- nodepool-zuul-functional http://logs.openstack.org/08/659708/2/check/nodepool-zuul-functional/52200f7/ : SUCCESS in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"99384028b1701b9d4a501a720f335d462d9e7f2f","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2019-05-17 06:09:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"efad0a085a69ff861c25102c988bba901dbb2106","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2019-05-17 06:39:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d72e947f1bb962bbd047018afe8371b2ae55675e","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2019-05-17 06:55:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e7c2dcc299dfab1c4a85bbb55b2c28e7909f2fda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 07:31:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- zuul-build-image http://logs.openstack.org/08/659708/5/check/zuul-build-image/b231a5d/ : SUCCESS in 18m 40s\n- tox-docs http://logs.openstack.org/08/659708/5/check/tox-docs/a393b40/html/ : SUCCESS in 4m 08s\n- tox-pep8 http://logs.openstack.org/08/659708/5/check/tox-pep8/463b6f3/ : SUCCESS in 4m 41s\n- tox-py35 http://logs.openstack.org/08/659708/5/check/tox-py35/cb04571/ : SUCCESS in 35m 16s\n- tox-py36 http://logs.openstack.org/08/659708/5/check/tox-py36/83b7080/ : SUCCESS in 33m 49s\n- nodejs-npm-run-lint http://logs.openstack.org/08/659708/5/check/nodejs-npm-run-lint/a871ec4/ : SUCCESS in 4m 16s\n- zuul-stream-functional-2.5 http://logs.openstack.org/08/659708/5/check/zuul-stream-functional-2.5/82fb562/ : SUCCESS in 8m 46s\n- zuul-stream-functional-2.6 http://logs.openstack.org/08/659708/5/check/zuul-stream-functional-2.6/68cdedc/ : SUCCESS in 5m 55s\n- zuul-stream-functional-2.7 http://logs.openstack.org/08/659708/5/check/zuul-stream-functional-2.7/9d01389/ : SUCCESS in 5m 50s\n- zuul-stream-functional-2.8 http://logs.openstack.org/08/659708/5/check/zuul-stream-functional-2.8/97606ca/ : SUCCESS in 6m 56s\n- zuul-tox-remote http://logs.openstack.org/08/659708/5/check/zuul-tox-remote/b10c8f5/ : SUCCESS in 18m 28s\n- zuul-quick-start http://logs.openstack.org/08/659708/5/check/zuul-quick-start/1462705/ : POST_FAILURE in 2m 13s\n- nodepool-zuul-functional http://logs.openstack.org/08/659708/5/check/nodepool-zuul-functional/15a70b6/ : SUCCESS in 9m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ef883bbaf780c1bde71dc5033d940813fae7adb4","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2019-05-17 07:32:16.000000000","message":"Patch Set 5:\n\nrecheck quick-start ipv6","accounts_in_message":[],"_revision_number":5},{"id":"1d3cb6d599a555f8c84a6d2f1a9f78f0410e19bd","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2019-05-17 07:56:58.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"52fc6e82f0c8ed8d418e433794f628ab44166336","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 08:37:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- zuul-build-image http://logs.openstack.org/08/659708/6/check/zuul-build-image/5103530/ : SUCCESS in 31m 01s\n- tox-docs http://logs.openstack.org/08/659708/6/check/tox-docs/26fd920/html/ : SUCCESS in 4m 11s\n- tox-pep8 http://logs.openstack.org/08/659708/6/check/tox-pep8/c9bee56/ : SUCCESS in 3m 43s\n- tox-py35 http://logs.openstack.org/08/659708/6/check/tox-py35/a8e07ba/ : SUCCESS in 37m 06s\n- tox-py36 http://logs.openstack.org/08/659708/6/check/tox-py36/47f2822/ : SUCCESS in 32m 52s\n- nodejs-npm-run-lint http://logs.openstack.org/08/659708/6/check/nodejs-npm-run-lint/74571fd/ : SUCCESS in 4m 37s\n- zuul-stream-functional-2.5 http://logs.openstack.org/08/659708/6/check/zuul-stream-functional-2.5/28fdbf9/ : SUCCESS in 6m 25s\n- zuul-stream-functional-2.6 http://logs.openstack.org/08/659708/6/check/zuul-stream-functional-2.6/425f76c/ : SUCCESS in 6m 39s\n- zuul-stream-functional-2.7 http://logs.openstack.org/08/659708/6/check/zuul-stream-functional-2.7/a0128b4/ : SUCCESS in 7m 31s\n- zuul-stream-functional-2.8 http://logs.openstack.org/08/659708/6/check/zuul-stream-functional-2.8/e8e02ad/ : SUCCESS in 6m 14s\n- zuul-tox-remote http://logs.openstack.org/08/659708/6/check/zuul-tox-remote/44c068f/ : SUCCESS in 16m 16s\n- zuul-quick-start http://logs.openstack.org/08/659708/6/check/zuul-quick-start/f00f396/ : SUCCESS in 7m 16s\n- nodepool-zuul-functional http://logs.openstack.org/08/659708/6/check/nodepool-zuul-functional/d58b6d2/ : SUCCESS in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5331daccea71fa01f51d416ee2b07c82d6ae44ea","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-05-17 12:13:46.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"31ac1eda190b3842588c2158f5f8aa86fb3d9fc1","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2019-05-17 12:29:22.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3f207ce5229b3d92290fa1a99d7da778276e0f2f","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2019-05-17 12:37:51.000000000","message":"Topic changed from tox_remote_debug to tox-remote","accounts_in_message":[],"_revision_number":6},{"id":"52de7d78d5fbef5720a1d3c3a575e059beca4406","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 12:42:59.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cec5cb305fbc40bfa6bec18745c22d4d916606f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 13:40:26.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- zuul-upload-image http://logs.openstack.org/08/659708/6/gate/zuul-upload-image/c4175bf/ : SUCCESS in 26m 21s\n- tox-docs http://logs.openstack.org/08/659708/6/gate/tox-docs/6c3fc1d/html/ : SUCCESS in 5m 12s\n- tox-pep8 http://logs.openstack.org/08/659708/6/gate/tox-pep8/dd5122a/ : SUCCESS in 4m 20s\n- tox-py35 http://logs.openstack.org/08/659708/6/gate/tox-py35/4d7ddb8/ : FAILURE in 35m 46s\n- tox-py36 http://logs.openstack.org/08/659708/6/gate/tox-py36/67849a9/ : FAILURE in 41m 28s\n- nodejs-npm-run-lint http://logs.openstack.org/08/659708/6/gate/nodejs-npm-run-lint/38c57d0/ : SUCCESS in 5m 25s\n- zuul-stream-functional-2.5 http://logs.openstack.org/08/659708/6/gate/zuul-stream-functional-2.5/1fd5c20/ : SUCCESS in 6m 43s\n- zuul-stream-functional-2.6 http://logs.openstack.org/08/659708/6/gate/zuul-stream-functional-2.6/a29ed21/ : SUCCESS in 6m 15s\n- zuul-stream-functional-2.7 http://logs.openstack.org/08/659708/6/gate/zuul-stream-functional-2.7/48a9ad8/ : SUCCESS in 6m 28s\n- zuul-stream-functional-2.8 http://logs.openstack.org/08/659708/6/gate/zuul-stream-functional-2.8/348dd3f/ : SUCCESS in 6m 35s\n- zuul-tox-remote http://logs.openstack.org/08/659708/6/gate/zuul-tox-remote/3c2da41/ : SUCCESS in 18m 17s\n- zuul-quick-start http://logs.openstack.org/08/659708/6/gate/zuul-quick-start/d05984e/ : POST_FAILURE in 2m 50s","accounts_in_message":[],"_revision_number":6},{"id":"0f6f04a6c6510e4cfba06953bfc046f31b4d05df","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2019-05-17 19:19:03.000000000","message":"Patch Set 6:\n\nrecheck ipv6 docker","accounts_in_message":[],"_revision_number":6},{"id":"fe32deb39a958a1bea4cba1c4191fa6e0dc53b6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 20:11:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- zuul-build-image http://logs.openstack.org/08/659708/6/check/zuul-build-image/2570f03/ : SUCCESS in 24m 55s\n- tox-docs http://logs.openstack.org/08/659708/6/check/tox-docs/0cc10ee/html/ : SUCCESS in 4m 18s\n- tox-pep8 http://logs.openstack.org/08/659708/6/check/tox-pep8/feba2fe/ : SUCCESS in 5m 32s\n- tox-py35 http://logs.openstack.org/08/659708/6/check/tox-py35/2678f0d/ : SUCCESS in 38m 33s\n- tox-py36 http://logs.openstack.org/08/659708/6/check/tox-py36/2e5217c/ : SUCCESS in 38m 21s\n- nodejs-npm-run-lint http://logs.openstack.org/08/659708/6/check/nodejs-npm-run-lint/bc26cf1/ : SUCCESS in 7m 05s\n- zuul-stream-functional-2.5 http://logs.openstack.org/08/659708/6/check/zuul-stream-functional-2.5/df7573e/ : SUCCESS in 6m 18s\n- zuul-stream-functional-2.6 http://logs.openstack.org/08/659708/6/check/zuul-stream-functional-2.6/aae9505/ : SUCCESS in 6m 48s\n- zuul-stream-functional-2.7 http://logs.openstack.org/08/659708/6/check/zuul-stream-functional-2.7/c5ba729/ : SUCCESS in 6m 32s\n- zuul-stream-functional-2.8 http://logs.openstack.org/08/659708/6/check/zuul-stream-functional-2.8/338a813/ : SUCCESS in 9m 15s\n- zuul-tox-remote http://logs.openstack.org/08/659708/6/check/zuul-tox-remote/a020371/ : SUCCESS in 22m 31s\n- zuul-quick-start http://logs.openstack.org/08/659708/6/check/zuul-quick-start/c1163b2/ : SUCCESS in 7m 37s\n- nodepool-zuul-functional http://logs.openstack.org/08/659708/6/check/nodepool-zuul-functional/393b0b8/ : SUCCESS in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f6d051ed0bac19f0625d639053fe3569b4ad4276","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 20:11:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4426236ea0c914576e7bd4bbada38fb2404e7ece","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 21:08:22.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- zuul-upload-image http://logs.openstack.org/08/659708/6/gate/zuul-upload-image/2b4e3da/ : SUCCESS in 27m 47s\n- tox-docs http://logs.openstack.org/08/659708/6/gate/tox-docs/97e11f1/html/ : SUCCESS in 4m 33s\n- tox-pep8 http://logs.openstack.org/08/659708/6/gate/tox-pep8/22a7449/ : SUCCESS in 4m 14s\n- tox-py35 http://logs.openstack.org/08/659708/6/gate/tox-py35/058ca2d/ : SUCCESS in 42m 39s\n- tox-py36 http://logs.openstack.org/08/659708/6/gate/tox-py36/fa73f97/ : SUCCESS in 37m 37s\n- nodejs-npm-run-lint http://logs.openstack.org/08/659708/6/gate/nodejs-npm-run-lint/55e3948/ : SUCCESS in 5m 25s\n- zuul-stream-functional-2.5 http://logs.openstack.org/08/659708/6/gate/zuul-stream-functional-2.5/1f101c2/ : SUCCESS in 6m 36s\n- zuul-stream-functional-2.6 http://logs.openstack.org/08/659708/6/gate/zuul-stream-functional-2.6/08cb937/ : SUCCESS in 6m 50s\n- zuul-stream-functional-2.7 http://logs.openstack.org/08/659708/6/gate/zuul-stream-functional-2.7/60a06f8/ : SUCCESS in 7m 04s\n- zuul-stream-functional-2.8 http://logs.openstack.org/08/659708/6/gate/zuul-stream-functional-2.8/12b9093/ : SUCCESS in 6m 40s\n- zuul-tox-remote http://logs.openstack.org/08/659708/6/gate/zuul-tox-remote/27bbfca/ : SUCCESS in 17m 45s\n- zuul-quick-start http://logs.openstack.org/08/659708/6/gate/zuul-quick-start/ba321ee/ : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":6},{"id":"0f7bad74858d3daa33f71791a238f9161f7d75aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 21:08:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"b23a2bf727497c7cc63ed92b6aac39ef1ac9c0b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 21:09:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- zuul-promote-image http://logs.openstack.org/08/659708/6/promote/zuul-promote-image/1d5ab7c/ : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"053e6388cd762bb438c09d0432fe30989c3215e0","revisions":{"db0266ec01a85f975905446e1dc11c0c65f0cfdf":{"kind":"REWORK","_number":1,"created":"2019-05-17 03:28:46.000000000","uploader":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"ref":"refs/changes/08/659708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/zuul/zuul","ref":"refs/changes/08/659708/1","commands":{"Checkout":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/zuul/zuul refs/changes/08/659708/1"}}},"commit":{"parents":[{"commit":"d23b32c592b94d9c965853e9b0801adea00e766e","subject":"Merge \"Revert \"web: upgrade react and react-scripts to ^2.0.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/d23b32c592b94d9c965853e9b0801adea00e766e"}]}],"author":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 03:23:09.000000000","tz":0},"committer":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 03:25:05.000000000","tz":0},"subject":"zuul_stream: prevent incorrect task exit code detection","message":"zuul_stream: prevent incorrect task exit code detection\n\nThis change improves task exit detection to prevent skipping mangled output.\nThis change also ensure the zuul-tox-remote job is testing the speculative\nzuul_console service instead of the one setup by the job.\n\nChange-Id: If2e2324235a6608985168d606d95adeade36afe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/db0266ec01a85f975905446e1dc11c0c65f0cfdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/db0266ec01a85f975905446e1dc11c0c65f0cfdf"}]},"branch":"refs/heads/master"},"a3c0c8e3ec97d72e09e09bde03cf74d2ec4527e2":{"kind":"REWORK","_number":2,"created":"2019-05-17 04:47:33.000000000","uploader":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"ref":"refs/changes/08/659708/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/zuul/zuul","ref":"refs/changes/08/659708/2","commands":{"Checkout":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/zuul/zuul refs/changes/08/659708/2"}}},"commit":{"parents":[{"commit":"d23b32c592b94d9c965853e9b0801adea00e766e","subject":"Merge \"Revert \"web: upgrade react and react-scripts to ^2.0.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/d23b32c592b94d9c965853e9b0801adea00e766e"}]}],"author":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 03:23:09.000000000","tz":0},"committer":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 04:47:43.000000000","tz":0},"subject":"zuul-tox-remote: kill previously started zuul_console service","message":"zuul-tox-remote: kill previously started zuul_console service\n\nThis change ensures the zuul-tox-remote job is testing the speculative\nzuul_console service instead of the one setup by the job.\n\nChange-Id: If2e2324235a6608985168d606d95adeade36afe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/a3c0c8e3ec97d72e09e09bde03cf74d2ec4527e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/a3c0c8e3ec97d72e09e09bde03cf74d2ec4527e2"}]},"branch":"refs/heads/master"},"086799785aa57d08312a0a2f92fed478a3052a8d":{"kind":"REWORK","_number":3,"created":"2019-05-17 06:09:31.000000000","uploader":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"ref":"refs/changes/08/659708/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/zuul/zuul","ref":"refs/changes/08/659708/3","commands":{"Checkout":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/zuul/zuul refs/changes/08/659708/3"}}},"commit":{"parents":[{"commit":"a0c299341137890b65be4d7f20ba72f45f699ac2","subject":"Merge \"tox: Integrate tox-docker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/a0c299341137890b65be4d7f20ba72f45f699ac2"}]}],"author":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 06:08:14.000000000","tz":0},"committer":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 06:08:14.000000000","tz":0},"subject":"zuul-tox-remote: use unique zuul_console service","message":"zuul-tox-remote: use unique zuul_console service\n\nThis change ensures the zuul-tox-remote job is testing the unique zuul_console\nservice instead of the one setup by the job.\n\nChange-Id: If2e2324235a6608985168d606d95adeade36afe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/086799785aa57d08312a0a2f92fed478a3052a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/086799785aa57d08312a0a2f92fed478a3052a8d"}]},"branch":"refs/heads/master"},"bfaf5157481fddb6c847671788dc76d780a22e63":{"kind":"REWORK","_number":4,"created":"2019-05-17 06:39:11.000000000","uploader":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"ref":"refs/changes/08/659708/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/zuul/zuul","ref":"refs/changes/08/659708/4","commands":{"Checkout":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/zuul/zuul refs/changes/08/659708/4"}}},"commit":{"parents":[{"commit":"a0c299341137890b65be4d7f20ba72f45f699ac2","subject":"Merge \"tox: Integrate tox-docker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/a0c299341137890b65be4d7f20ba72f45f699ac2"}]}],"author":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 06:08:14.000000000","tz":0},"committer":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 06:39:20.000000000","tz":0},"subject":"zuul-tox-remote: use unique zuul_console service","message":"zuul-tox-remote: use unique zuul_console service\n\nThis change ensures the zuul-tox-remote job is testing the unique zuul_console\nservice instead of the one setup by the job.\n\nChange-Id: If2e2324235a6608985168d606d95adeade36afe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/bfaf5157481fddb6c847671788dc76d780a22e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/bfaf5157481fddb6c847671788dc76d780a22e63"}]},"branch":"refs/heads/master"},"40bf522f54af9dc84badd7bd7b717c8f0ae8d216":{"kind":"REWORK","_number":5,"created":"2019-05-17 06:55:28.000000000","uploader":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"ref":"refs/changes/08/659708/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/zuul/zuul","ref":"refs/changes/08/659708/5","commands":{"Checkout":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/zuul/zuul refs/changes/08/659708/5"}}},"commit":{"parents":[{"commit":"a0c299341137890b65be4d7f20ba72f45f699ac2","subject":"Merge \"tox: Integrate tox-docker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/a0c299341137890b65be4d7f20ba72f45f699ac2"}]}],"author":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 06:08:14.000000000","tz":0},"committer":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 06:55:44.000000000","tz":0},"subject":"zuul-tox-remote: use unique zuul_console service","message":"zuul-tox-remote: use unique zuul_console service\n\nThis change ensures the zuul-tox-remote job is testing the unique zuul_console\nservice instead of the one setup by the job.\n\nChange-Id: If2e2324235a6608985168d606d95adeade36afe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/40bf522f54af9dc84badd7bd7b717c8f0ae8d216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/40bf522f54af9dc84badd7bd7b717c8f0ae8d216"}]},"branch":"refs/heads/master"},"053e6388cd762bb438c09d0432fe30989c3215e0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-05-17 07:56:58.000000000","uploader":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"ref":"refs/changes/08/659708/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/zuul/zuul","ref":"refs/changes/08/659708/6","commands":{"Checkout":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/659708/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/zuul/zuul refs/changes/08/659708/6"}}},"commit":{"parents":[{"commit":"91663cd23813f394cf14244cbcc95aa250039731","subject":"Fix race in log streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/91663cd23813f394cf14244cbcc95aa250039731"}]}],"author":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 06:08:14.000000000","tz":0},"committer":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2019-05-17 07:56:58.000000000","tz":0},"subject":"zuul-tox-remote: use unique zuul_console service","message":"zuul-tox-remote: use unique zuul_console service\n\nThis change ensures the zuul-tox-remote job is testing the unique zuul_console\nservice instead of the one setup by the job.\n\nChange-Id: If2e2324235a6608985168d606d95adeade36afe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/053e6388cd762bb438c09d0432fe30989c3215e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/053e6388cd762bb438c09d0432fe30989c3215e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
