)]}'
{"id":"zuul%2Fzuul~680310","triplet_id":"zuul%2Fzuul~master~I62d4e783e94528126cd4a7d85b3e664e84758bf1","project":"zuul/zuul","branch":"master","topic":"pagure-driver-update","hashtags":[],"change_id":"I62d4e783e94528126cd4a7d85b3e664e84758bf1","subject":"Pagure - handle initial comment change event","status":"MERGED","created":"2019-09-05 08:01:20.000000000","updated":"2019-09-18 15:01:50.000000000","submitted":"2019-09-18 14:59:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680310-1568818772556-d2cb535d","meta_rev_id":"19f30167953752539984814e0dfeeaeee4a40c21","_number":680310,"virtual_id_number":680310,"owner":{"_account_id":6889,"name":"Fabien Boucher","email":"fboucher@redhat.com","username":"fabien-boucher"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-09-18 14:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-09-18 14:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-09-18 10:29:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-09-18 14:09:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 15:55:33.000000000","updated_by":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"reviewer":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"state":"REVIEWER"},{"updated":"2019-09-18 10:29:05.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-09-18 14:09:05.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-09-18 14:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9bd7e1056806de96b07e7a23277318dbec4306d","author":{"_account_id":6889,"name":"Fabien Boucher","email":"fboucher@redhat.com","username":"fabien-boucher"},"date":"2019-09-05 08:01:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1aae65f794277d448ec2c8aec78f45bb0eebcd1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 09:00:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- zuul-build-image https://zuul.opendev.org/t/zuul/build/1690d32a89914cc5b6cd94efbf91d3a4 : SUCCESS in 47m 50s\n- zuul-tox-docs https://zuul.opendev.org/t/zuul/build/353db82d05d247fa9e402b2cc72f1af0 : SUCCESS in 5m 35s\n- tox-pep8 https://zuul.opendev.org/t/zuul/build/af4b504874c04357852e1aab8b427798 : SUCCESS in 5m 18s\n- tox-py35 https://zuul.opendev.org/t/zuul/build/7e76b6ab66444938b30885b7388024d6 : SUCCESS in 39m 40s\n- tox-py36 https://zuul.opendev.org/t/zuul/build/5d4f58cfc70d4d74ad8caf423f90c137 : SUCCESS in 43m 54s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/zuul/build/c0e5812e80e54e268c38d043db94dfbc : SUCCESS in 5m 02s\n- zuul-tox-remote https://zuul.opendev.org/t/zuul/build/511eb0d88368414ca266d9a27678bbdd : SUCCESS in 21m 14s\n- zuul-quick-start https://zuul.opendev.org/t/zuul/build/09c60ecab98146448dd22e8e37ca2bcf : SUCCESS in 18m 37s\n- nodepool-zuul-functional https://zuul.opendev.org/t/zuul/build/c2a071436f1349b4bb68a039d5565f24 : SUCCESS in 11m 10s (non-voting)\n- zuul-build-python-release https://zuul.opendev.org/t/zuul/build/fdc471d8a9704698bc785e13ee426caf : SUCCESS in 5m 28s\n- build-javascript-content-tarball https://zuul.opendev.org/t/zuul/build/754a4d2889384e8f9253838e2dab4b03 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0fd2a640e6e65e7327a1e73c1c0a8727df03bd4e","author":{"_account_id":6889,"name":"Fabien Boucher","email":"fboucher@redhat.com","username":"fabien-boucher"},"date":"2019-09-10 10:06:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3f7b6e5c74ac6d6d67cd02b4715941326d4397e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 10:50:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- zuul-build-image https://zuul.opendev.org/t/zuul/build/532e773e7af748959ddd263799e20218 : SUCCESS in 37m 29s\n- zuul-tox-docs https://zuul.opendev.org/t/zuul/build/6522e03900cc4d38810e2a7b61e4ac04 : SUCCESS in 6m 48s\n- tox-pep8 https://zuul.opendev.org/t/zuul/build/85d19462cb824640a848041fbe040de2 : SUCCESS in 6m 25s\n- tox-py35 https://zuul.opendev.org/t/zuul/build/8a5355f65a3b45a0b4b1d23cfe0d1796 : SUCCESS in 37m 10s\n- tox-py36 https://zuul.opendev.org/t/zuul/build/aa2fe7527fed46c48b3aebe056a38515 : SUCCESS in 42m 08s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/zuul/build/dc0576119555460c9c7c525205f5f292 : SUCCESS in 5m 06s\n- zuul-tox-remote https://zuul.opendev.org/t/zuul/build/613a8d0ed847471d8ac7b99cec668f79 : SUCCESS in 19m 15s\n- zuul-quick-start https://zuul.opendev.org/t/zuul/build/73a25c6f8fec449db2cd07ffd4052334 : SUCCESS in 8m 42s\n- nodepool-zuul-functional https://zuul.opendev.org/t/zuul/build/0733b3e8ec68497b82e3f9de264232f8 : SUCCESS in 11m 48s (non-voting)\n- zuul-build-python-release https://zuul.opendev.org/t/zuul/build/3e0f01bd1ee84893adf950b95c275e9a : SUCCESS in 5m 07s\n- build-javascript-content-tarball https://zuul.opendev.org/t/zuul/build/1c1df3a373fb46dfbfee8faec3db3e22 : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f8eca3d3b7bac62163391dc43f3adc7300674a80","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-09-12 14:15:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a445b99cdb1fc807d443fad9367d4a63d1757677","author":{"_account_id":6889,"name":"Fabien Boucher","email":"fboucher@redhat.com","username":"fabien-boucher"},"date":"2019-09-18 10:02:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7b65c49eae0c5a719aac0574554c2adfcdc8d63c","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-09-18 10:29:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2c5b01d988a30e44b27f4d7514a1c05fc6acfb5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:57:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- zuul-build-image https://zuul.opendev.org/t/zuul/build/9f4152de89ee4760b69e9519cde820bc : SUCCESS in 33m 20s\n- zuul-tox-docs https://zuul.opendev.org/t/zuul/build/2a062dcf7f5d40bdae61102e14b8967e : SUCCESS in 5m 28s\n- tox-pep8 https://zuul.opendev.org/t/zuul/build/be2c4f4ede52463ab91d9f7afef473d9 : SUCCESS in 5m 28s\n- tox-py35 https://zuul.opendev.org/t/zuul/build/dac63db1a86c41709897b8eb68414652 : SUCCESS in 53m 31s\n- tox-py36 https://zuul.opendev.org/t/zuul/build/ace3b8d3319c4d1dbff2a1e0cc964068 : SUCCESS in 38m 54s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/zuul/build/f06c0f25aaa74fa4a96f29e88d696e3b : SUCCESS in 4m 36s\n- zuul-tox-remote https://zuul.opendev.org/t/zuul/build/49f92d929a9c4472a98bf7a88d59b474 : SUCCESS in 22m 38s\n- zuul-quick-start https://zuul.opendev.org/t/zuul/build/15d2890202a5497fa68d98a66f3606f5 : SUCCESS in 8m 20s\n- nodepool-zuul-functional https://zuul.opendev.org/t/zuul/build/89d7088e4e424c0487f01f0e91cb7ea7 : SUCCESS in 14m 37s (non-voting)\n- zuul-build-python-release https://zuul.opendev.org/t/zuul/build/bdad4a66113b4e7b84d18747c087ec0d : SUCCESS in 5m 16s\n- build-javascript-content-tarball https://zuul.opendev.org/t/zuul/build/1edd7596747849f79d83d1f7b8877968 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":3},{"id":"0ffeaaa694a04b5e4eb570ac496aae9a0097da86","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-09-18 14:09:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5ec2b1ecc1dddcb8670213d231cc093c94faac7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:09:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bec1ffe915e7af718058060b9f61c3b5f7602be9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:59:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded.\n\n- zuul-upload-image https://zuul.opendev.org/t/zuul/build/507dcec7134d46ffb23af5ad1bd436d3 : SUCCESS in 39m 46s\n- zuul-tox-docs https://zuul.opendev.org/t/zuul/build/63824dc772a94222bcfed1d466602394 : SUCCESS in 6m 12s\n- tox-pep8 https://zuul.opendev.org/t/zuul/build/9ed6cef582f44bf9afe4c39ef91cd0f8 : SUCCESS in 4m 14s\n- tox-py35 https://zuul.opendev.org/t/zuul/build/f85f4e74f02f46f69449a58a0386a51a : SUCCESS in 42m 01s\n- tox-py36 https://zuul.opendev.org/t/zuul/build/05804e76c7cf4368be2444048d650ec4 : SUCCESS in 48m 31s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/zuul/build/63cb32d0eab04163bd2dab54bbd26aae : SUCCESS in 6m 01s\n- zuul-tox-remote https://zuul.opendev.org/t/zuul/build/ef5b087f31e547869e80e08542483312 : SUCCESS in 23m 38s\n- zuul-quick-start https://zuul.opendev.org/t/zuul/build/92a3faa510ea498a9dfdb0689266130b : SUCCESS in 12m 39s\n- zuul-build-python-release https://zuul.opendev.org/t/zuul/build/f934da01ca91470fb6d288f1229312e9 : SUCCESS in 6m 37s\n- build-javascript-content-tarball https://zuul.opendev.org/t/zuul/build/8ca67d3d7b5a41c8a3db9132e3502eac : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7a46e4e7ee6d33247c50b7dd49be57c726b6b4cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:59:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"22092121e4b099b020ecb5ddda4470be3b021928","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 15:01:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- zuul-promote-image https://zuul.opendev.org/t/zuul/build/f5a91ffe8f3245ecb0d58ce6632dfff9 : SUCCESS in 1m 50s\n- zuul-promote-docs https://zuul.opendev.org/t/zuul/build/4b3f3ac8eeea4ccfb4d0c4ade2800ce0 : SUCCESS in 1m 28s\n- opendev-promote-python https://zuul.opendev.org/t/zuul/build/6360f03fc8334e548c151cb0cf9c6372 : SUCCESS in 1m 29s\n- opendev-promote-javascript-content https://zuul.opendev.org/t/zuul/build/447cb9debf3242b99c91c54c6e8c3f6f : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3c2b5c6ed2dd772d4d9e11d291abdf7710356867","revisions":{"bb19df0c43ec94fea8b15cd0b248cb045f502778":{"kind":"REWORK","_number":1,"created":"2019-09-05 08:01:20.000000000","uploader":{"_account_id":6889,"name":"Fabien Boucher","email":"fboucher@redhat.com","username":"fabien-boucher"},"ref":"refs/changes/10/680310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/zuul/zuul","ref":"refs/changes/10/680310/1","commands":{"Checkout":"git fetch https://review.opendev.org/zuul/zuul refs/changes/10/680310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/zuul/zuul refs/changes/10/680310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/zuul/zuul refs/changes/10/680310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/zuul/zuul refs/changes/10/680310/1"}}},"commit":{"parents":[{"commit":"73b31e7453377b1b2b0b67cb283815bd5b3c2ed4","subject":"Pagure - add support for git.tag.creation event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/73b31e7453377b1b2b0b67cb283815bd5b3c2ed4"}]}],"author":{"name":"Fabien Boucher","email":"fboucher@redhat.com","date":"2019-09-05 07:55:52.000000000","tz":120},"committer":{"name":"Fabien Boucher","email":"fboucher@redhat.com","date":"2019-09-05 07:55:52.000000000","tz":120},"subject":"Pagure - handle initial comment change event","message":"Pagure - handle initial comment change event\n\nThe \"initial comment\" is the first comment of a PR on Pagure. It\nis used to provide the Depends-on stanza. Recently Pagure added the\ncapability to send an event when that initial comment is changed:\nhttps://pagure.io/pagure/issue/4398\n\nThis change handles the event as a PR changed to retrigger the attached\njobs.\n\nChange-Id: I62d4e783e94528126cd4a7d85b3e664e84758bf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/bb19df0c43ec94fea8b15cd0b248cb045f502778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/bb19df0c43ec94fea8b15cd0b248cb045f502778"}]},"branch":"refs/heads/master"},"ca8f690aec2123b5c0f6e270d14610bf79157786":{"kind":"REWORK","_number":2,"created":"2019-09-10 10:06:00.000000000","uploader":{"_account_id":6889,"name":"Fabien Boucher","email":"fboucher@redhat.com","username":"fabien-boucher"},"ref":"refs/changes/10/680310/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/zuul/zuul","ref":"refs/changes/10/680310/2","commands":{"Checkout":"git fetch https://review.opendev.org/zuul/zuul refs/changes/10/680310/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/zuul/zuul refs/changes/10/680310/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/zuul/zuul refs/changes/10/680310/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/zuul/zuul refs/changes/10/680310/2"}}},"commit":{"parents":[{"commit":"73b31e7453377b1b2b0b67cb283815bd5b3c2ed4","subject":"Pagure - add support for git.tag.creation event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/73b31e7453377b1b2b0b67cb283815bd5b3c2ed4"}]}],"author":{"name":"Fabien Boucher","email":"fboucher@redhat.com","date":"2019-09-05 07:55:52.000000000","tz":120},"committer":{"name":"Fabien Boucher","email":"fboucher@redhat.com","date":"2019-09-10 10:05:45.000000000","tz":120},"subject":"Pagure - handle initial comment change event","message":"Pagure - handle initial comment change event\n\nThe \"initial comment\" is the first comment of a PR on Pagure. It\nis used to provide the Depends-on stanza. Recently Pagure added the\ncapability to send an event when that initial comment is changed:\nhttps://pagure.io/pagure/issue/4398\n\nThis change handles the event as a PR changed to retrigger the attached\njobs.\n\nChange-Id: I62d4e783e94528126cd4a7d85b3e664e84758bf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/ca8f690aec2123b5c0f6e270d14610bf79157786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/ca8f690aec2123b5c0f6e270d14610bf79157786"}]},"branch":"refs/heads/master"},"3c2b5c6ed2dd772d4d9e11d291abdf7710356867":{"kind":"REWORK","_number":3,"created":"2019-09-18 10:02:07.000000000","uploader":{"_account_id":6889,"name":"Fabien Boucher","email":"fboucher@redhat.com","username":"fabien-boucher"},"ref":"refs/changes/10/680310/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/zuul/zuul","ref":"refs/changes/10/680310/3","commands":{"Checkout":"git fetch https://review.opendev.org/zuul/zuul refs/changes/10/680310/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/zuul/zuul refs/changes/10/680310/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/zuul/zuul refs/changes/10/680310/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/zuul/zuul refs/changes/10/680310/3"}}},"commit":{"parents":[{"commit":"795520d28616b3756d4533856aa01d1cc8fa697e","subject":"Pagure - fix wrong commit gitweb url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/795520d28616b3756d4533856aa01d1cc8fa697e"}]}],"author":{"name":"Fabien Boucher","email":"fboucher@redhat.com","date":"2019-09-05 07:55:52.000000000","tz":120},"committer":{"name":"Fabien Boucher","email":"fboucher@redhat.com","date":"2019-09-18 09:49:12.000000000","tz":120},"subject":"Pagure - handle initial comment change event","message":"Pagure - handle initial comment change event\n\nThe \"initial comment\" is the first comment of a PR on Pagure. It\nis used to provide the Depends-on stanza. Recently Pagure added the\ncapability to send an event when that initial comment is changed:\nhttps://pagure.io/pagure/issue/4398\n\nThis change handles the event as a PR changed to retrigger the attached\njobs.\n\nChange-Id: I62d4e783e94528126cd4a7d85b3e664e84758bf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/3c2b5c6ed2dd772d4d9e11d291abdf7710356867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/3c2b5c6ed2dd772d4d9e11d291abdf7710356867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
