)]}'
{"id":"zuul%2Fzuul~707708","triplet_id":"zuul%2Fzuul~master~Iebd6dafd9e040b6c443ebe1ed1e2702d37804399","project":"zuul/zuul","branch":"master","hashtags":[],"change_id":"Iebd6dafd9e040b6c443ebe1ed1e2702d37804399","subject":"Report robot comments to gerrit","status":"MERGED","created":"2020-02-13 19:02:48.000000000","updated":"2020-02-13 22:31:43.000000000","submitted":"2020-02-13 22:28:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707708-1581632910398-3390009c","meta_rev_id":"e568f8b651e1e35c1d826ef1c38a6fda3c6c6bf6","_number":707708,"virtual_id_number":707708,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 22:28:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-02-13 21:35:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-02-13 19:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-02-13 21:35:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 19:19:39.000000000","updated_by":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"reviewer":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"state":"REVIEWER"},{"updated":"2020-02-13 21:35:31.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-02-13 22:28:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df9464e391ca9aaef5e8981d30a902c1a44a42e8","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-02-13 19:02:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"caa57c1618682f3a634c877e2e16029ce470f52d","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2020-02-13 19:19:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"75a48d211637845857a90f8ba8495ef901ccebd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 20:01:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- zuul-build-image https://zuul.opendev.org/t/zuul/build/9fa9560511384752903e6032747cd7df : SUCCESS in 37m 45s\n- zuul-tox-docs https://zuul.opendev.org/t/zuul/build/458f7b92138b4162a3f45c7a05ee454a : SUCCESS in 10m 41s\n- tox-pep8 https://zuul.opendev.org/t/zuul/build/f7339c6bd3ba4afbbc59176a11943463 : SUCCESS in 10m 49s\n- tox-py35 https://zuul.opendev.org/t/zuul/build/1a9fed9bdadf4ebca118e6f520edf93b : SUCCESS in 46m 52s\n- tox-py37 https://zuul.opendev.org/t/zuul/build/7fcfe7c0a96842d695242510789a0846 : SUCCESS in 51m 58s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/zuul/build/af3bda476d28468db8ec995b134e4b00 : SUCCESS in 10m 14s\n- zuul-tox-remote https://zuul.opendev.org/t/zuul/build/b9487441243749d08a97233a87dbfc39 : SUCCESS in 23m 19s\n- zuul-quick-start https://zuul.opendev.org/t/zuul/build/500383ce49cf4ea6bc621376268cf9ef : SUCCESS in 10m 23s\n- nodepool-zuul-functional https://zuul.opendev.org/t/zuul/build/359cb00d4e2843b0921b7bbd2f799b7f : SUCCESS in 17m 35s (non-voting)\n- zuul-build-python-release https://zuul.opendev.org/t/zuul/build/505a5643f9d34c9bafe67839750ef5c3 : SUCCESS in 6m 55s\n- build-javascript-content-tarball https://zuul.opendev.org/t/zuul/build/f3f8c608f4c141a493b305e326470246 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":1},{"id":"fb3e4839862010597e79fea8f017541836ddaf63","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-02-13 21:35:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"191ac713fda6751468b35f9dc7b6e866aaafd4f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 21:35:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4b3961bedaac66517438bd6c38189f993ddfc37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 22:28:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- zuul-upload-image https://zuul.opendev.org/t/zuul/build/f5ddd7dab33c4eeab6555054c97cb432 : SUCCESS in 33m 36s\n- zuul-tox-docs https://zuul.opendev.org/t/zuul/build/70bef4839e9248789826259c924da481 : SUCCESS in 6m 39s\n- tox-pep8 https://zuul.opendev.org/t/zuul/build/51c6af59bf0b462db1adf103bf696bd8 : SUCCESS in 8m 07s\n- tox-py35 https://zuul.opendev.org/t/zuul/build/65c382a93a7d4ec79c81b4c5bfaa307a : SUCCESS in 49m 35s\n- tox-py37 https://zuul.opendev.org/t/zuul/build/b55ae888902e495f8ffe370bb8f7898e : SUCCESS in 40m 57s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/zuul/build/22ab8f0163184dc886ec75dc57ad7f08 : SUCCESS in 8m 26s\n- zuul-tox-remote https://zuul.opendev.org/t/zuul/build/f309becfe2cf49649915972781c85fa3 : SUCCESS in 23m 12s\n- zuul-quick-start https://zuul.opendev.org/t/zuul/build/145db37547a6450aa0a9e6efddd3d7fc : SUCCESS in 8m 50s\n- zuul-build-python-release https://zuul.opendev.org/t/zuul/build/4999eb948c454698894405a6625fc46e : SUCCESS in 9m 33s\n- build-javascript-content-tarball https://zuul.opendev.org/t/zuul/build/77546c31c0a1409b8360c1e8821ba544 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"30193023f82282c9339be07faec39bb9fe529ff9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 22:28:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e568f8b651e1e35c1d826ef1c38a6fda3c6c6bf6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 22:31:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- zuul-promote-image https://zuul.opendev.org/t/zuul/build/b67cc8b8e0bd44b6a93c19602d923dbb : SUCCESS in 2m 49s\n- zuul-promote-docs https://zuul.opendev.org/t/zuul/build/96751cc0434f45489a1c2d5853afc3a0 : SUCCESS in 1m 52s\n- opendev-promote-python https://zuul.opendev.org/t/zuul/build/8580780a4e984112ac2747ccaf0cc9b9 : SUCCESS in 1m 35s\n- opendev-promote-javascript-content https://zuul.opendev.org/t/zuul/build/c7ba2f7d067c49b0821d26c4b97f451a : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1b95639b9c6a4a277419f7c5047495608fd7dc9","revisions":{"d1b95639b9c6a4a277419f7c5047495608fd7dc9":{"kind":"REWORK","_number":1,"created":"2020-02-13 19:02:48.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/08/707708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/zuul/zuul","ref":"refs/changes/08/707708/1","commands":{"Checkout":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/707708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/707708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/zuul/zuul refs/changes/08/707708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/zuul/zuul refs/changes/08/707708/1"}}},"commit":{"parents":[{"commit":"d48be299ab25990b484f3097d0cc74622fca484a","subject":"Merge \"Don\u0027t report enqueue of non-live item\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/d48be299ab25990b484f3097d0cc74622fca484a"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2020-02-13 19:00:14.000000000","tz":-480},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2020-02-13 19:02:41.000000000","tz":-480},"subject":"Report robot comments to gerrit","message":"Report robot comments to gerrit\n\nIf there are line comments, we are supposed to always report them.\nHowever, when we added support for robot comments, we didn\u0027t update\nthe check which determined whether to post a review.  We do that\nif at least one of the following is true: there is a message, there\nis a label, there is a line comment.  That should also be true if\nthere is a robot comment (since that is a form of line comment).\n\nThis updates that check and adds a test.\n\nChange-Id: Iebd6dafd9e040b6c443ebe1ed1e2702d37804399\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/d1b95639b9c6a4a277419f7c5047495608fd7dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/zuul/zuul/commit/d1b95639b9c6a4a277419f7c5047495608fd7dc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
